)]}'
{"id":"openstack%2Frequirements~786572","triplet_id":"openstack%2Frequirements~master~Ia9849db7c74c03545971afb8ecc68168abf6d704","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Ia9849db7c74c03545971afb8ecc68168abf6d704","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-04-16 06:33:29.000000000","updated":"2021-04-16 18:32:46.000000000","submitted":"2021-04-16 18:31:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786572-openstack/requirements/constraints","meta_rev_id":"f620d9555602694c9c525567109fb129c59a8307","_number":786572,"virtual_id_number":786572,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-16 18:31:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-16 14:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-16 14:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-16 06:36:39.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-04-16 07:08:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-16 08:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ee72ddd2159c0be934a72b710fc55fb70862a62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-16 06:33:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f632ba45fd0759822937557d7589d0939f9caf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-16 06:36:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"137a965a39c83f0eaea7838675ea01951015ae43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 07:08:22.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d860518d35fc4d429c4f0a8870993e58 : FAILURE in 17m 43s","accounts_in_message":[],"_revision_number":2},{"id":"200a782520a57395511dbd831a20fb9e8150fe67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 08:09:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864577f8eb224695b865fc96b4a405be : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec93f2f8a3d449718ba9f7f0a7bf70b1 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cad93fe039549d39b25c611ef2386a3 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc7f6020311140319c9de42b6c0364c7 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77d8c8ea68d24df1978f1f83a837f49a : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/68b3b37ff86248c3b4f596b518122c09 : SUCCESS in 6m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f550392ee1eb41e1a1b16ec1a996473c : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f09ac771b8eb43bfb77ffd6ef1a98d1a : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9ed26d8262e04c5fa4b3195083564533 : SUCCESS in 9m 12s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cd0bd6c2fa7246ff9995e3cc8218bc0c : SUCCESS in 18m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/798b9b6389b84da381d6ca8808c98085 : SUCCESS in 11m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6cb0c8e628094b75b0361d0240328987 : SUCCESS in 8m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fcf792bff7a24625a6eccb30efb461c2 : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2eaaf536bb38403e84be855a5f2e48bd : SUCCESS in 11m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/86f0d730ec914c56a86c6d90a0ea7fe5 : SUCCESS in 16m 06s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f8c478ba78bb45cba2a317a589ab2494 : SUCCESS in 7m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0b579734ff384dbb9300239092c47aa3 : SUCCESS in 26m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/e280b88fff3b4dc589161c3b652735d5 : SUCCESS in 12m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c426c3938bc4e0588631213d38c7146 : SUCCESS in 7m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/19fb9b43469f469690ae246a456a592b : SUCCESS in 19m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09b9677a6b5e49e195f87605b2f82b08 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7cd72ba2a8f449d8b421c48ac0edf5fb : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f9cce4f10b24476fa1a707e9d242627e : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6148ed5898184725861d8b4e8876a1b2 : SUCCESS in 38m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9abac8247ebf4d4a9eb6adfb0da18237 : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":2},{"id":"27f220871a050dd109d65a111d7bbbe36e093884","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-16 14:23:57.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"edfe3a5a329e660cfdf9612433337f44414e1ae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 14:26:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8322d5b08fa4c3059e4b586d501076b82607547d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 18:31:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e2ce8c046f464b833687cef70cd189 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75fb52a9cb754d99b05a2c6e8988ea78 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1171441e09a4519af7ad55c5d66463b : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/445ab6b69054469cb34e8b2cd32b50fe : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/847ecb1969754f38a6d12d0bbd16adbd : SUCCESS in 5m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0545f598dce74077b3518019ad212ccb : SUCCESS in 5m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f3ec456f310453c8d10e10eb0597a10 : SUCCESS in 14m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da47e0044d634ce8a8ce211acca0423a : SUCCESS in 11m 31s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/266ebf8fba9340dcb1de21c7165605c6 : SUCCESS in 15m 50s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/aa186f4853004f9eabd0c11ee574a452 : SUCCESS in 7m 20s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7b2c2975ae42445c9874e6274e82ebb3 : SUCCESS in 15m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/506bdbec2cb64f4fafdfa87b40e6e482 : SUCCESS in 9m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3f4b358bcddc49339e2bc01375c366bc : SUCCESS in 12m 37s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/60b1152c5a8d4b89863dac51bbd917a2 : SUCCESS in 16m 43s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c834215aa45f474fb1cc41b4490de402 : SUCCESS in 5m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d630fb1713e04f9cb11579e4797addb6 : SUCCESS in 28m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/804f995492434229bd614c7e2c5aa622 : SUCCESS in 13m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bbb94875653346d1ba10ebce5bfa0d60 : SUCCESS in 4m 19s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a71c328d56f4ed7a595bd44f0356a88 : SUCCESS in 18m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5e58f8b4c45047d4bb8ffde089138b6e : SUCCESS in 6m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bab391c9d0a345a0b2c833497cdb5fcd : SUCCESS in 7m 25s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c6df8c4fb9c5491f8fda2ef436ead62c : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ac6156460c6428fa787432f1d73e747 : SUCCESS in 1h 00m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1df31d6633744c80897a597f0dc28221 : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":2},{"id":"21f71dc42921190b8b1a4cdba4ecd9413b27bda3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 18:31:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f620d9555602694c9c525567109fb129c59a8307","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-16 18:32:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b03099cc0114dec866088b23c00010f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","revisions":{"6e8f87590153ad3bc7db2995e946efa10c386fea":{"kind":"REWORK","_number":1,"created":"2021-04-16 06:33:29.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/72/786572/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/786572/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/786572/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/786572/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/786572/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/786572/1"}}},"commit":{"parents":[{"commit":"3539e11b97fba2f8d94a451799bea52f4ba601ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3539e11b97fba2f8d94a451799bea52f4ba601ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-16 06:33:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-16 06:33:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia9849db7c74c03545971afb8ecc68168abf6d704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e8f87590153ad3bc7db2995e946efa10c386fea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6e8f87590153ad3bc7db2995e946efa10c386fea"}]},"branch":"refs/heads/master"},"cd565a13d7fb8e635023f22cd05e9e2539ac2048":{"kind":"REWORK","_number":2,"created":"2021-04-16 06:36:39.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/72/786572/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/72/786572/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/786572/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/786572/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/72/786572/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/72/786572/2"}}},"commit":{"parents":[{"commit":"3539e11b97fba2f8d94a451799bea52f4ba601ac","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3539e11b97fba2f8d94a451799bea52f4ba601ac"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-16 06:33:27.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-04-16 06:36:29.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ia9849db7c74c03545971afb8ecc68168abf6d704\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
