)]}'
{"id":"openstack%2Frequirements~787100","triplet_id":"openstack%2Frequirements~master~Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-04-20 06:29:48.000000000","updated":"2021-04-26 18:07:11.000000000","submitted":"2021-04-26 18:05:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"787100-openstack/requirements/constraints/noclob","meta_rev_id":"3032e67ecfc508d16110189a42c1c785139dba42","_number":787100,"virtual_id_number":787100,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 18:05:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-26 16:46:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-26 16:46:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 06:53:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-20 09:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 07:11:59.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-04-26 15:08:11.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f64a93c06af7a05fb070e7991392d4c2e11645cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-20 06:29:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4beacf7370619e36149d4849ca983a4d40b4838f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 06:53:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a52e0774982d4b7abfec1055c89ff5fe : FAILURE in 17m 22s","accounts_in_message":[],"_revision_number":1},{"id":"22d9c39a0e7d8c53ee86340d5703ebb89ec669a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 09:15:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4adaf4606f6d4eecb495cf605d5783d1 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f679bc3dc46f4e25a6789e47bf4e57d4 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e610b076ddc4dd585c22e2771b4cff9 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cefaf76984c84e42a38a4886eed953be : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/423c0548a64c431ba1ae6792d3aee61a : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/673fa297407e4832887dfb993c5caac4 : SUCCESS in 8m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6754971721b54ea592d9194e740abbe4 : SUCCESS in 6m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4ba5607e2fda41d8b93a5701c4ff08fc : FAILURE in 7m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f08a60cd54c04b7eaaab4ece8b9bd1b2 : FAILURE in 6m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6243e1b618ab4f9f8d09653575726bd9 : FAILURE in 11m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/645b1cf052164cb6b2fa6362f64f2533 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1e24c3ec625043a3ae407abbe0202038 : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/776b9074b27d4591b7755c40b86485f5 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6dcfd63e10a747089a65a43c8e7fe0c9 : FAILURE in 12m 49s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c02e637d48764cefa97b8e4b3b1fd70e : FAILURE in 14m 30s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9d8b4b48503d411f92508a5dab01d1e3 : FAILURE in 6m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/66516bd4478145d488eaeea262510334 : FAILURE in 33m 33s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1103e85193364146a98959cd448757f5 : FAILURE in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c4afe59c90d84dc7a56fd2a42a29a600 : SUCCESS in 6m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97438fd5dc464189a7f6972c6794cb91 : FAILURE in 18m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a8df5995bd58477eaea7aea756a49331 : FAILURE in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/178082fc854245a48d5a5ea4bb462f34 : SUCCESS in 6m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c56c36e3ea574b66a55802c132bd9021 : SUCCESS in 12m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb3bfea30a0d4344bff7c6990b269c18 : FAILURE in 1h 57m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c7686b7a0a04dbf88637f959bf2ee27 : TIMED_OUT in 2h 03m 30s","accounts_in_message":[],"_revision_number":1},{"id":"d51c293bed4692d6b0746fb008278b1b2ce04c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-21 06:22:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7d024cbbb5eb6ccd420bf17104e944f63fc75b3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 07:15:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/616f433c314548d0afd093d822c1a44a : FAILURE in 10m 10s","accounts_in_message":[],"_revision_number":2},{"id":"7ca6bf40c9601131d22fd4f95282b1de5fa3529c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 09:12:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc8e8acfadf4ff6b725824a49f3a1e5 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4c3385beb14a9a9503e3cee4d4d33b : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f856a6a9a8847f289992392b16af58f : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af8e5448ea54d90971591ee8f9c15ab : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/033e2425b0d14431b5d43d76e225ec57 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8de3de2e42e4740885d75fa25879311 : SUCCESS in 5m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72674f45765e44faacb2164bf00c956b : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8831bfb31ae4138a49c015597ae998f : FAILURE in 8m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e16bfc748914b988e3304a40e91b7b3 : FAILURE in 5m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/199380331a1945aa93d6f7879bfd6794 : FAILURE in 11m 02s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c4da4cb8341349bc85a8e9677c711478 : SUCCESS in 5m 57s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/fbb4c5ae9b4f4d6aacab9778e2367cf1 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9c8f4dd97c444518a695e5e1ef1abc4f : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aaae8b6531d34efcb41cda0eb6a6bb47 : POST_FAILURE in 12m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0fd495d1a0df41a8bdf76a96602dcf32 : FAILURE in 12m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/e971d43ec85a4a3da5068747a3429aa8 : FAILURE in 7m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/091f8d2735204f7aa747e45dbdf619d0 : FAILURE in 30m 14s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6e7b2e02703747dc92d214b721fa4171 : FAILURE in 12m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4fd8b88d683e4f03ae26575f4f02bfc7 : SUCCESS in 6m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/633939b69d15457c938e2f59ccb0bd63 : FAILURE in 20m 01s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e8d6dd5d30bb455fa232e955ea3b3808 : FAILURE in 6m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a54ce72e46af4e5bb909c5e5148248f3 : SUCCESS in 9m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4d2a16ed1d4347748e28581a7f5caa60 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6b75e531f9543b88a977f05886c7292 : FAILURE in 2h 05m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5a67b64eb6344fbcb5de7d11eb1b3615 : TIMED_OUT in 2h 03m 07s","accounts_in_message":[],"_revision_number":2},{"id":"283f6d2883938e0ade428280eb4ad4737128080d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-22 06:29:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dd9a98ca53a279529992476370ff34cf09265eaf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 06:49:46.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ef432523961343208d2c81900b6ea30f : FAILURE in 12m 00s","accounts_in_message":[],"_revision_number":3},{"id":"a09b1464416503f265846f03335d69851a3f9fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:00:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f248a92925b48d2963d396862b7c7ca : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1f1e897ca54312855650fa3a55f52a : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f28fc51cfcba43ddaf31e5fccb3547ae : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8caf17c4531e4cc7a18fe549a23f251a : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fdc3ea47fad4e1382b51206968738ae : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66d4136617464ae3944070b37674c2f9 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f356408322be4303a2e110d55b10f749 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/31bd3984d1c241279aab01b82fc2de98 : FAILURE in 7m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4caa9a0ab26b44b3822e024f2e31d2c1 : FAILURE in 5m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/82270bfc993b4d939e450ca0a74e2d67 : FAILURE in 9m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/edcc7c4ec5834103ac8369dac0d87a55 : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/84fc21c8a8bd483693420da12729e0fe : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8192e790b674a87b97dbca1cd348c9d : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/aedbdddaee7542e2a66068523a4d4915 : FAILURE in 11m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/de259814b8f64b5aa71d4be548800cef : FAILURE in 13m 36s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a497207b827141a1a4846f420aaf6ffb : FAILURE in 5m 32s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/324b3c5a49b2459293a4eff2390d2a83 : FAILURE in 37m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ded8415ebaa4e54adfe64b6126519a7 : FAILURE in 13m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/414c7a64c1f14575a5eca9d71dd98a6b : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f29d5a0659294931a8d9d8561213b32f : FAILURE in 22m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/513bb29699844898997ed5191535a19f : FAILURE in 6m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ea15be0ab4f426fa020f094c3d8a2a8 : SUCCESS in 7m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/685e6faff85b4faaac3c3b635869a6d6 : SUCCESS in 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9954acd2d2d84cbcb90871901f8aa1dc : FAILURE in 2h 11m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fb8e2476fef46b9bccfbf3d6c521d14 : TIMED_OUT in 2h 03m 35s","accounts_in_message":[],"_revision_number":3},{"id":"27617b7b80a7f6325020906fc763e33aa8dc95a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-23 06:36:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"67c36460e8cb1df108f7da7a18b27eb5a2b49a82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 06:47:32.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ca21bbcce9d54c16ba3daaff6c468b3a : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":4},{"id":"a4d6a03005c467a71cacc9842fc506a84cfaa58a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-04-23 07:11:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11d417f20cced78f890bdf9926be6e8d747d718d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:45:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec2122dee9f14d31bcc8438a12323c9a : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f328bb30d3741dba104a7f3f84e990f : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af06254ce739455eaf1b093fcc2014c2 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/298c86d860784d1583c72867e6334a02 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f84b5bdff324ec1aacf86592765a1d8 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/21bc6b18c8c24996872e17dc2fc89230 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4a948dae6fce49078eb3e92b81446cba : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d867628eb6ab46f894727302f9fa7a56 : FAILURE in 8m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f60cf880cdc047ec8de27cf4ad648409 : FAILURE in 7m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cc1f2df6305842859736e129816dc67b : FAILURE in 10m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a44a3b5ebb024b0c80cc1dda6b15088e : SUCCESS in 6m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/779627140ca54e1fb5b315ed6bfe82ff : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dbd8ecefece048e2b24ff9f04fb33c0c : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/786fc9e49b9749c18c4baa924dc44b33 : FAILURE in 9m 52s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8785230bb49f41b9841fd817860d73d4 : FAILURE in 13m 00s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/58576f5654994856800f6b89a64befd7 : FAILURE in 6m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/2581ebcb0a9d4390973348127215d694 : FAILURE in 35m 24s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d4da3e08c59e4a6b9d02a1709d3bf0a4 : FAILURE in 12m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/de8b38bdf51743da968917e5aee297f6 : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2dcf3a5d9dd84a9b983c1b2c6fcaf8e7 : FAILURE in 19m 20s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/60b76c8f90c34829be0a1c619ae0d88f : FAILURE in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/100099baf54e4aeab4bee9837541069d : SUCCESS in 7m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/95b498c6ec594357aa14a12b1dfcab61 : SUCCESS in 20m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ebd7e227e054c7097b69e46898c9d71 : FAILURE in 2h 09m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c6e9fddad77a4c4ab054b0b6d1ac9b0f : TIMED_OUT in 2h 03m 37s","accounts_in_message":[],"_revision_number":4},{"id":"db2a29a47db804179a8f1a52b908bfb136d5a0fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-24 06:46:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5cfa0030df318e5931e7689bec129bfc922a4c8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-24 07:02:29.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/66088d8a98f14160afc50380631f38a4 : FAILURE in 8m 42s","accounts_in_message":[],"_revision_number":5},{"id":"e6f7fa5635b4f2edd1e01aa9c738fb304acbcdc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-24 08:54:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13f2c6561f284f88a8fde0b8709533b5 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f124535543f4e2cbaa121174d5fd235 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7aeb3cb30d5e4d94bb115ed333b40a16 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7764ab37a19f402ead84f6f0b36e5e5c : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bc31a7e7ce04309a0e967b420438d3e : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f30c06b0d15b402287b4860b44f56c56 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9a9d31c07cdd4236a8e867a6c690d9a4 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b352e95923014e0882b2fc47cf7a3839 : FAILURE in 4m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/20917efd9f1c442c8e8f50df468d5e9a : FAILURE in 4m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3a7d1fca0c47491eafcc4c77154dea04 : FAILURE in 9m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cac23037d487438c9ab8837f92fab504 : SUCCESS in 8m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/dfcf6eee9b43471ab9357e38adba5257 : SUCCESS in 7m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/979305d85d1244eab30da9474903f78c : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d79371b94b2149ce87c3b7c956173ed7 : FAILURE in 11m 19s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b0684f8200784387ab7b35b101182218 : FAILURE in 11m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/62f9d5b8b50e425c95d892a2cb8048e7 : FAILURE in 4m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/81bf59e06f30494ea00632b078b655ec : FAILURE in 29m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1669574953ab4fd69e33e01a6ee8157b : FAILURE in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a8aeb33b56a40aa825eaf3e4bc630ac : SUCCESS in 6m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eaf9d5453994e5399ba0fd4b610d26c : FAILURE in 18m 39s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5d001326e0ad4373b64338d8063b2e6b : FAILURE in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba0307ee135c46958f8a1fc04c321f46 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8271defe7b7a4d4db66a86b8b77ef682 : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f6f5f736129a4da394ca521dc738cdcf : FAILURE in 2h 01m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4860e1c2a91943e9b5abc99d26c3afb6 : TIMED_OUT in 2h 03m 03s","accounts_in_message":[],"_revision_number":5},{"id":"f9ac6f647f8aace928691ccafadb8469e0b4792b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-26 06:33:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4a5ea06b9699d1f070b4356628f385a5b53606d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 06:50:47.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/914ba43d18d64bcaa1f1ce0b16c72c26 : FAILURE in 13m 30s","accounts_in_message":[],"_revision_number":6},{"id":"bccd2e86cf0c1f946549266763d3a7b66279f256","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:00:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc36f2622574091a9e34024a6c74690 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf1920c40d147e8b0f8ef9ca3470c9d : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f5909efa2df43fdb930fae4e3358f01 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05c1a16aa0684496a9160985e09050d3 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69a4846f1f3442ab9ffc98da8d9218d2 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/faa29a891e124bdf99696f07dd3c26a8 : SUCCESS in 6m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/31efdf367df049d48b86dced519a8981 : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/71a0ac47b2db4716a5e31791095f36eb : FAILURE in 5m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/20304b563a074f29bfc471a43dde9ea8 : FAILURE in 7m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b17a6ead644d4ddd98916d2c279f232d : FAILURE in 10m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8d2b7e02a4dd4feb918f7606e5659361 : SUCCESS in 7m 07s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6a314509c14f4abb9578543d5300e869 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63ad2be43eee4611b0ea68633c324963 : SUCCESS in 7m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e03cec9b664c43c4bfa8e7144ccdd542 : FAILURE in 11m 28s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9a55418d8dad4371a4d3af23465af80a : FAILURE in 12m 44s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b59f9d4deca94c368e590f8eee51e068 : FAILURE in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/51a08b544a4f4ee9906adbfaac9acc62 : FAILURE in 27m 29s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fec4c634c10f4d8182758abbf9ca7faa : FAILURE in 11m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/177ee2462dc049a78676489826799f8b : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45643f82100749c287690fdf1d724117 : FAILURE in 18m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/23f98b4a9e6b419ca64a639f0edb93fe : FAILURE in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cf51819c544b4da282c28ef3002c8c81 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ffd6c4ffd43b4a9a952470703488a7c3 : SUCCESS in 13m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8717149ef944883833edcc4f69cec21 : FAILURE in 2h 07m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfe2b3ae22df4744b8e9b3141605bbc4 : TIMED_OUT in 2h 03m 30s","accounts_in_message":[],"_revision_number":6},{"id":"3747e4dff782ba3b7fee8e72a08d0560313588dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-26 15:08:11.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"01fee3a270dd14fa2db542b110c51882198cd1db","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-26 15:08:30.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":7},{"id":"c1577fca0b2aa0f4d36f0dad21bed65336bd34a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 15:26:09.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8c1f1b1b92df4c12ae078aa9fb0af9b9 : FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":7},{"id":"d988f1f64a901a01b2bb10b0c3c6bcd1704b6299","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:25:40.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d50b10bab442f8b479609084172dd4 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89664334cbce42ef9a736d45a032af0c : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/316d6bb914ab46baa16a21f03facad5c : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6551c90e79e64bfa82d2a7dad245e716 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22b7fad4eeb8479fa322a9a4095bffb2 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5eb6b59139841bcb80004678b8e7454 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc96d782f3314814979e188d9e0b8f5a : SUCCESS in 4m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88acad8615ae4d97b24f5927ce75043d : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/37d09d8b113f48f1b4ee02ae42fcf663 : SUCCESS in 10m 01s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/226a9258bf6d4a04ad7c3c60a60e3a79 : SUCCESS in 10m 35s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/821bafc637e0400291cb22fc08cc25cf : SUCCESS in 6m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c1f482b6296945c5b03342526e877948 : SUCCESS in 9m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a29fafb346764cbe9b4826562aba8517 : SUCCESS in 8m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b3c36e6830f140bc89853f28eb8b7c1d : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74f3dd7ce0514d39a2b1694e2e729f0c : SUCCESS in 15m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/155a45b6bed046f0828b82d0836dd292 : SUCCESS in 4m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b792f9e273c54964a4ae8e3ee1cdd93c : SUCCESS in 29m 11s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1c062ce7256f4a2a9383791936d07530 : SUCCESS in 12m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba8bb3e53d9e40e18e6da6cff7e19581 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d04459680e504e9cafe992e9f636b329 : SUCCESS in 17m 54s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/347381e30ebe46fbaeb30442fd36aa89 : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71b4f76e4ea54ea4aa75f57bf9b88589 : SUCCESS in 7m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2b957447087e42e692a9dc8043aa3833 : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/107d3f79b29b448ab8d9c29a84a60711 : SUCCESS in 47m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6e933e9fd401454398932cdc192bb326 : SUCCESS in 1h 13m 06s","accounts_in_message":[],"_revision_number":7},{"id":"4e0cc4e93d8dbb340088a9e2bb2451c9b48dfe1d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-26 16:46:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1eebab59941e81f4fc2480b0f00e9d1c5a71689e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:46:53.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"65df0f9b41ce2218fb1933471c99c0163f46dbaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:05:51.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed952145d674696bfe3dc461365e17e : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822a4b4fe7164ebabc51463f1e3e76d1 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da1d581aad2c4275ae8e899e985571f1 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce473b7a958f4daa857d77c7e5cb351a : SUCCESS in 6m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9dfab4b79a3411480168c5a4768f832 : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a7cf0146783542cdb82312502205c2c6 : SUCCESS in 4m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c6eb9086ea4248f9b9bbd11eb3997ee7 : SUCCESS in 16m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/80446c24ab4e4628a6bbd0abd1d6e91b : SUCCESS in 10m 40s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/510fa77f95104bbc871cec593c01c678 : SUCCESS in 9m 52s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/779d4ea138d3411f87f3c91d1af405d1 : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f08311ef3e8f488e98f41dfaa74ade7c : SUCCESS in 8m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b221190985f0412e8ef26ca88ff1f81b : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0476ccc96ea14e1888658ad1ef55a137 : SUCCESS in 10m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/65de9ae6f029485db4f50631408c9d2a : SUCCESS in 15m 41s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f5d70b708a724bbe9a5eef4939e96b07 : SUCCESS in 6m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f150bb89f948490ca37ff344b471a076 : SUCCESS in 30m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a18a1d026cee4db6a0370bd72a1b4443 : SUCCESS in 11m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d04eef82f8fa4cefa858e6e0e892541c : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/023c69549d9e42cc8471eb794ecf2a15 : SUCCESS in 16m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/aa4a8993e24f4937872cc3e24576c02e : SUCCESS in 7m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f3ae18446e6452d9e3beb2b6edc9575 : SUCCESS in 7m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8d82b158ece5415ab91cea9aabbdcf4f : SUCCESS in 9m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26d76099cfe84d1baba08ed550de3146 : SUCCESS in 48m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e99fe03bd02e43d7969be3032a7c3093 : SUCCESS in 1h 17m 08s","accounts_in_message":[],"_revision_number":7},{"id":"cf05bdf4e1af4d1ef054cc43d4d645a3811d3b54","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:05:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"3032e67ecfc508d16110189a42c1c785139dba42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 18:07:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/347a9d0dfa294874a8c5ead07baba5e7 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"4793be53f6fc3b4c3f81ec8b4860793678d8a746","revisions":{"3e907ee7ac739dfc7b372ea2db28329c3879cf7b":{"kind":"REWORK","_number":1,"created":"2021-04-20 06:29:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/787100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/1"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-20 06:29:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-20 06:29:46.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e907ee7ac739dfc7b372ea2db28329c3879cf7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e907ee7ac739dfc7b372ea2db28329c3879cf7b"}]},"branch":"refs/heads/master"},"7bc08e9dba49276eb92c0063bd7674804fdfa238":{"kind":"REWORK","_number":2,"created":"2021-04-21 06:22:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/787100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/2"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-21 06:22:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-21 06:22:04.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bc08e9dba49276eb92c0063bd7674804fdfa238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7bc08e9dba49276eb92c0063bd7674804fdfa238"}]},"branch":"refs/heads/master"},"cb0e434112cdc7533434d0d9264cad12c1731ce1":{"kind":"REWORK","_number":3,"created":"2021-04-22 06:29:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/787100/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/3"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-22 06:29:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-22 06:29:52.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb0e434112cdc7533434d0d9264cad12c1731ce1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb0e434112cdc7533434d0d9264cad12c1731ce1"}]},"branch":"refs/heads/master"},"3c977fa82fd8f81455dffa9aa0ea6db8d5fd6451":{"kind":"REWORK","_number":4,"created":"2021-04-23 06:36:25.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/787100/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/4"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-23 06:36:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-23 06:36:23.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c977fa82fd8f81455dffa9aa0ea6db8d5fd6451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3c977fa82fd8f81455dffa9aa0ea6db8d5fd6451"}]},"branch":"refs/heads/master"},"08b83adf086eb929617a6eb7b480a3ff6e571159":{"kind":"REWORK","_number":5,"created":"2021-04-24 06:46:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/787100/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/5"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-24 06:46:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-24 06:46:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08b83adf086eb929617a6eb7b480a3ff6e571159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08b83adf086eb929617a6eb7b480a3ff6e571159"}]},"branch":"refs/heads/master"},"55de9ecbe3db9ffd72e71f30e61158fb03473641":{"kind":"REWORK","_number":6,"created":"2021-04-26 06:33:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/787100/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/6"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 06:33:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 06:33:24.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55de9ecbe3db9ffd72e71f30e61158fb03473641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55de9ecbe3db9ffd72e71f30e61158fb03473641"}]},"branch":"refs/heads/master"},"4793be53f6fc3b4c3f81ec8b4860793678d8a746":{"kind":"REWORK","_number":7,"created":"2021-04-26 15:08:11.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/00/787100/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/787100/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/787100/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/787100/7"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 06:33:24.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-04-26 15:07:27.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ic98e7b9b2f7eeb15a2eabc2792ba9b0792277d71\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4793be53f6fc3b4c3f81ec8b4860793678d8a746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4793be53f6fc3b4c3f81ec8b4860793678d8a746"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
