)]}'
{"id":"openstack%2Frequirements~787951","triplet_id":"openstack%2Frequirements~master~Ia6e31031801db629e156ff40e81a1289e0601e8b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia6e31031801db629e156ff40e81a1289e0601e8b","subject":"update constraint for neutron-lib to new release 2.11.0","status":"MERGED","created":"2021-04-26 10:12:20.000000000","updated":"2021-04-26 14:25:13.000000000","submitted":"2021-04-26 14:23:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787951-new-release","meta_rev_id":"186b277cb1ff6171d1446560764d14e5e056caf4","_number":787951,"virtual_id_number":787951,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 14:23:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-04-26 10:55:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-04-26 12:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-04-26 11:19:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-04-26 12:04:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 10:27:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 10:55:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-04-26 10:55:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-04-26 11:19:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-26 11:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 12:04:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"4ff22b33c6f871a9e78d586af704d4597154ed78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-26 10:12:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"488f61bd8529fa0898b0a7d7c95e5ff2c49b9d47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:27:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c5a0a269d5f344bdaa57cdca25725a66 : FAILURE in 10m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a7a4cada64e5e0555cf52c64597609cfdfac2e62","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-04-26 10:55:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f202d00d677990ff5966b7ad1d261cdaa1991c7f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-26 11:19:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edfff4ebf82525a0011730c4053058caba3736a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 11:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3046e98f3e184e56a5075381b44f6a3c : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb990792ffac414a8e354a760df5caa8 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08385a634761456a99ef071b0f4efb0d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb985ed3e1d041fea0d60773eb6c5c5c : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b89c7e9761c45d5a258c6dd7a2a47a0 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d21394e28a5347c4941666b66663f082 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e4a2b6a1c14146fab0c0674e6b9e7680 : SUCCESS in 5m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f660dad85ab4c13927b082006961306 : SUCCESS in 10m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/25883115b3a7419ebe7642f7b3671eb6 : SUCCESS in 10m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9d93d0b60b5d420ba31cf774afe205f1 : SUCCESS in 11m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/8e65dd25e1e346a1942e3e1012fb7c72 : SUCCESS in 6m 19s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/30fa4c4d1f7545c0be6bbcadc536f133 : SUCCESS in 11m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/74a56a69466b43cfa6f83888f50e71d7 : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/64520091babc40d0a484dd9a0616df9c : SUCCESS in 10m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/310074b81a2e440da1eda734e40ce76c : SUCCESS in 14m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2c60651c4d4a40b8a28d5f753ab1c1c2 : SUCCESS in 6m 14s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d3f4cdc4970e4cd8804e43575069c981 : SUCCESS in 28m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/44a9b618485c4063bb9ca44486dd9598 : SUCCESS in 16m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2477cc442b5c41689e24907e610d5d34 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14b4474e68c541ceb7062846795ab0bd : SUCCESS in 18m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a4da47c15da946fca1037cc09fc1f206 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/871ce2d3e6474d45b63528d06dfdc43d : SUCCESS in 7m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4cfccb42656d4105b94c022446e869e1 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36c08332ff4d4553a689fa33fc35e9c1 : SUCCESS in 56m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/205b387acdac472cb6f365f807e3b453 : SUCCESS in 1h 08m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cad5759f2b2161da1c17c0d675f2df53cbcb922f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-04-26 12:04:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"836ea9a0ae3cf487901cccd40b9f20118bfad842","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:04:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0af3da996a1aeb01ad4bea5ef310c47a477e5854","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:23:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbf978a11860484881783326420b7192 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af2ae866ef44b4b87d525bd784d9416 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fc8ac1ea1c146e18ab4f4c53fcae130 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31bdd8b3573d45449dd6dc2c269b7a1b : SUCCESS in 6m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e9e6982f64e42858440db43d2789fa3 : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/39a41df254d444ffa2ca813d61f9e57e : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba01d03e0811471ba28c94bb9cacd293 : SUCCESS in 9m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d19f162bb3c648c2bf7209a993622f72 : SUCCESS in 9m 29s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4085cfea42b84e969fd7628f9d7e2bd6 : SUCCESS in 9m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6e06ec462e57451ba2974d993dd02586 : SUCCESS in 6m 33s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/598f7c2502694d83b260cc351602c566 : SUCCESS in 10m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b9587b0888f4ae39435aad79b05046f : SUCCESS in 9m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/969031bf441a45068b736bd1fa57d3ec : SUCCESS in 12m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d0b61f2d33cf44b2bd87086f9f2aaba4 : SUCCESS in 13m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/6e5299a31220489d8b1457221697666e : SUCCESS in 7m 03s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/5b413c138ed942b6b77d04d9fa267fa6 : SUCCESS in 26m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/86d1808b43424a1aa326e0e70187bb62 : SUCCESS in 13m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/52ce2e3aefa84fbfaf7d73313e9613fd : SUCCESS in 5m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d402f4db3dcb480d934e6937f795e920 : SUCCESS in 16m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/444e83e307dd4c6a9d3bb70752a4613d : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e49ab0b066ce4e2c9b773686a32031f3 : SUCCESS in 7m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fababea56af946f98325e0a122c8287b : SUCCESS in 13m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61f58d00dc444a778b4d2018909c9c6d : SUCCESS in 49m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/950d7eb4115347b29e77f10df4e035c3 : SUCCESS in 1h 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8b88ec7b7110172c407b19122e2192331142e0e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:23:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"186b277cb1ff6171d1446560764d14e5e056caf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:25:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d37073bb87ba49e6ae25b0f3a0874c41 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"beb486c35dde9fa6fd884a0057fac1754b92c701","revisions":{"beb486c35dde9fa6fd884a0057fac1754b92c701":{"kind":"REWORK","_number":1,"created":"2021-04-26 10:12:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/51/787951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/51/787951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/787951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/787951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/51/787951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/51/787951/1"}}},"commit":{"parents":[{"commit":"cd565a13d7fb8e635023f22cd05e9e2539ac2048","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cd565a13d7fb8e635023f22cd05e9e2539ac2048"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 10:12:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 10:12:17.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.11.0","message":"update constraint for neutron-lib to new release 2.11.0\n\nmeta: version: 2.11.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: I58b718c5a97f0cf61df4fb7640fa0a7c1eb6131b\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia6e31031801db629e156ff40e81a1289e0601e8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/beb486c35dde9fa6fd884a0057fac1754b92c701"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/beb486c35dde9fa6fd884a0057fac1754b92c701"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
