)]}'
{"id":"openstack%2Frequirements~788016","triplet_id":"openstack%2Frequirements~master~Ifb092290664a38724c84e70ffb22eb0020cffdb6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ifb092290664a38724c84e70ffb22eb0020cffdb6","subject":"update constraint for openstacksdk to new release 0.56.0","status":"MERGED","created":"2021-04-26 14:47:47.000000000","updated":"2021-04-26 19:16:00.000000000","submitted":"2021-04-26 19:13:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788016-new-release","meta_rev_id":"ed91d221cabcae324473e402f97568e6aae38935","_number":788016,"virtual_id_number":788016,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 19:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-26 17:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-26 17:33:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 15:00:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-26 16:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 17:33:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"dbce7ff3c1d01e83c8ce9408fded8997fce6a256","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-26 14:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84f7cd0ccf3428c0d34e3e2e1bee5e318040d74f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 15:00:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/532c9b49dc314fce8d950e2376d28026 : FAILURE in 9m 39s","accounts_in_message":[],"_revision_number":1},{"id":"842c8f01a0a881ebaff33897367a4e5388c069d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:10:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e08dfaf1025443e9fb25b434d4527cf : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce3d384b235142ea888fe09539d63e6a : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0f2588f23d34f4d9f326537e2ee8f59 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad7257759bdb4efb8df8a055c0144a96 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7fa51866f0b4978a2ab49b0c2955599 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e53ba75a7894bb1998c3f0b12653fc8 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/467364b8a5dc45ce92a2e8e3321a3f9c : SUCCESS in 3m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/db094c0c16d445a2990c531e2cfb915d : SUCCESS in 9m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/878539a539b44226a600743efe06b326 : SUCCESS in 9m 35s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2de8f5f813134b59ba062056ae922593 : SUCCESS in 9m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ef05f8038a504cd784528df44c2405af : SUCCESS in 5m 26s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/25a5ab82a0ed441daab6a0562f884ffa : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8155a1e4dae5419fb290277abb11f09e : SUCCESS in 6m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/00348020cb0243da8af94cbc9aee14c8 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70913e0387f74d9bbb3c4b08d2ae8d62 : SUCCESS in 13m 02s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a9a6d9c8e31d46f087698ba3cc684925 : SUCCESS in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/0045e4a9a934489aac5bdb38aeb9cc82 : SUCCESS in 28m 35s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/a4f300e59e224d7f9461595fc6cb2a1b : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac412788b8f240309ce5ce2764eb763e : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc47bd69bdce42018991966a79e4b88b : SUCCESS in 18m 36s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/030b6537e35e4f8a96479cc965bcc92f : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/07a34e10f9a449e48733b57d0dac3c57 : SUCCESS in 7m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71f0152c13b54a7a8fd437aee9f285d1 : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca33b7f6737847a08adbb43547dedb78 : SUCCESS in 49m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9370bbc5ff6472885bcce32d537f726 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b75b96e2f9e34c3349e560d1ff979d64525d8816","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-26 17:33:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b49eb7e4a36a692a5acbc37b131923cfe4125836","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:34:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13514656c80d6de4c5c3d7ed1e3d2c4aa8e60805","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:13:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07698dc213054cdaa71668745ed04062 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ef2a6b3a5f4ca8afad94274988bd48 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5606c6df238448a5b03790cc4f8237ed : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31d00c97e0cd41ab9805a8ba10319a45 : SUCCESS in 6m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d4c4e80ffb4241d2b40fa5eb425bcb05 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b032fe90ce248fcb694a4edf1c5630e : SUCCESS in 5m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/95fe947d52904971babf905cb44e3d14 : SUCCESS in 12m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b48e7748087f488b994bae2a7d4bc281 : SUCCESS in 19m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9129f8f26d4b441898489f30827164ae : SUCCESS in 12m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fc08cb64d9474e07b3f9b0ddaa0d734e : SUCCESS in 6m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c191ad62534c43f58e90923c3bc0342b : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/906c77dc1cf44c878f7058fe2e49b88e : SUCCESS in 10m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b6034ad0135f4839a953a79249e57dc7 : SUCCESS in 13m 32s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/3f71c6f44ebb491f846f34f0bfe9a7c5 : SUCCESS in 14m 18s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d55be659770442d38375b305d6d09499 : SUCCESS in 6m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/4b2d4a5787c14c4795d5768c6b3571cd : SUCCESS in 33m 38s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/85085cba726e40389f47efd22ce1647a : SUCCESS in 16m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b82e30a3eb93434c9afa2ea0937e64c8 : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e9bf3c7e0b4436eb7316dec05974ead : SUCCESS in 24m 16s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a51ed0e30e80448886686185286cbff2 : SUCCESS in 6m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f22adb4f8e241a880e00411c7b58f91 : SUCCESS in 9m 40s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d96bb4df551f4653b0d6c170f1ec431b : SUCCESS in 39m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9f4f27fa9d74a78ac500923bf99bdc4 : SUCCESS in 59m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b55ab13e14d249848f7695fdfafd123a : SUCCESS in 1h 26m 04s","accounts_in_message":[],"_revision_number":1},{"id":"f9d0b7098ad3f4d166666c5f51670b7a0c420754","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:13:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ed91d221cabcae324473e402f97568e6aae38935","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:16:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58d586a42f074ce4828ac7104b4a7b7e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a5c80a52265931156de34b1b2ab490a0d6335ba4","revisions":{"a5c80a52265931156de34b1b2ab490a0d6335ba4":{"kind":"REWORK","_number":1,"created":"2021-04-26 14:47:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/16/788016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/16/788016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/788016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/788016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/16/788016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/16/788016/1"}}},"commit":{"parents":[{"commit":"beb486c35dde9fa6fd884a0057fac1754b92c701","subject":"update constraint for neutron-lib to new release 2.11.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/beb486c35dde9fa6fd884a0057fac1754b92c701"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 14:47:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-26 14:47:45.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.56.0","message":"update constraint for openstacksdk to new release 0.56.0\n\nmeta: version: 0.56.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Commit: Jens Harbott \u003cj.harbott@x-ion.de\u003e\nmeta: release:Change-Id: Icc386a9fe36ddae4edbc15603e745b14d1861271\nmeta: release:Code-Review+1: Dr. Jens Harbott \u003cj.harbott@x-ion.de\u003e\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ifb092290664a38724c84e70ffb22eb0020cffdb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5c80a52265931156de34b1b2ab490a0d6335ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a5c80a52265931156de34b1b2ab490a0d6335ba4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
