)]}'
{"id":"openstack%2Frequirements~788341","triplet_id":"openstack%2Frequirements~master~I68d31601a337847ce646ac250521fb8809938014","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I68d31601a337847ce646ac250521fb8809938014","subject":"update constraint for oslo.config to new release 8.6.0","status":"MERGED","created":"2021-04-27 18:39:46.000000000","updated":"2021-04-28 19:01:16.000000000","submitted":"2021-04-28 18:58:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788341-new-release","meta_rev_id":"116857de6c3335265deb226b789f6b6fc8069b83","_number":788341,"virtual_id_number":788341,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 18:58:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-04-27 19:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-04-28 15:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-04-28 09:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-04-28 15:01:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 18:53:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-27 19:16:39.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-04-27 20:03:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-28 09:01:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-04-28 15:01:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ad13cfc59af8b729174fa21b396e8e8a15ed2452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-27 18:39:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dea6aefa761ebd12f686d957c127faaa3d91806","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:53:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/30040b299c89407faa298c72c92e1ed2 : FAILURE in 9m 40s","accounts_in_message":[],"_revision_number":1},{"id":"e506c4ae02c0831410d7554f57b56c2cce76d5eb","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-04-27 19:16:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"407319293a87b7bda505d4ca1d4dd927095e0ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 20:03:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a7ba47c8a304299a9ff08722dcb0546 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032a855e19294810ae23d3e32b33465c : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/565109cfa6fe4127a40c8c31526c063c : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3275745ce3c48629d3bcdd90ea50801 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5d8cc2abe3e48159eb63f8c3d50314c : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f6a38444c1724dd8a616acb3b7436934 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b7326b2e9bd4ff3940c5b24b9f335a0 : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da0364d944d94f26a78f50629ce0c39e : SUCCESS in 9m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5b3a2963a2b64b27b0c8cc21884b5004 : SUCCESS in 9m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2e8f4cf3148c4a36a8cd31a28af2df09 : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f8c0fda2a61a48e483b5e03dc749d83c : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/ed5dc389e537410d9a7f2afedd5f895a : SUCCESS in 10m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5610f53a6d364d829bdc43f9d336d3a4 : SUCCESS in 6m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b37eb5f12d734cbe95b6e78d166cb6a5 : SUCCESS in 9m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8921cf92201e43629efcb8950fce2d99 : SUCCESS in 15m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b6314cf795ae4cbc8d05d50f62e4d0d0 : SUCCESS in 5m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f8365d5a7b374e9692552aa28f7502ba : TIMED_OUT in 40m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/dad632f3a160410193e17101cae33cc2 : SUCCESS in 12m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b42ebca161b04ddcb42d66166cb90fae : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a89899453d654a5893b1a882217e3002 : SUCCESS in 17m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/18e54d8a475247ca8fd328211f5e6cfd : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/969af66116094e6e8121ea08fa65eade : SUCCESS in 6m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/741949ff55714121adfba5fa18b8e02a : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9154f06b7dcf4c5490b97db84f5c4d04 : SUCCESS in 56m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56ed81f98e6c40a3b0aaf7a8aeead4fd : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8875d8037e519ea6f257ec4ec30f17c7ab6444e4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-04-28 09:01:13.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aca2120d6836acb95aeb23c23b312c6e69fef64d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 09:16:50.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/737ee5fc3e88497790e7381b047a5630 : FAILURE in 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d48b2689079d3ee162d185366a67321cf0db97da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 10:16:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a497673b108c4a6ab7a6cded85b70ec9 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6abac2a4384b0ba8243949934e7562 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5eceaf9f093d4ba5930a66f93fe5d17a : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a32f782be72d41edac03d1729aba27b8 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be55ee82da4c4524824b831832727a16 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a270d11293d945ef86a1361aedbb7a36 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d2540b17584498ea269a2d0b8e3c9e6 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a38fd33665984fc3bb41102713afe47a : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e06bfcbf42424f26ad2aa0a9f2aace82 : SUCCESS in 9m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a2c969b260dd4c009379d217755e6b5a : SUCCESS in 9m 54s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0df1218f9a8e46e3a1a1fe1da7aa5b3a : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/24c732c9efeb42c4b05e8a5712a09194 : SUCCESS in 11m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/113113c1322240d79720c787b211ae89 : SUCCESS in 7m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3214709fcece4c599e8144f1c4d0407e : SUCCESS in 10m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5702efc6e1914463bf0077508a6a768b : SUCCESS in 13m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/64d2af4b57a34ae9962823b1d04770b8 : SUCCESS in 5m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bd09b96bd4c945cf822ff5cae49687ab : SUCCESS in 29m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/99c7dd461eb241e28a541f9fc845a6a3 : SUCCESS in 12m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/74a169e2018f4c6e826761d1fcad4c6e : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7f64937360314668b63e6277ad40b281 : SUCCESS in 17m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57ab484d3f204482b5351cdce56bb177 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/841a5150efd44b8bb36d857dd031d248 : SUCCESS in 8m 15s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6e5afc2a95114579b6f4d76a456d9cc5 : SUCCESS in 13m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22529e34e6764685ab9ddadf78af774a : SUCCESS in 54m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f253c13e6bc344a7b254d14fac32a441 : SUCCESS in 1h 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ff9355703272b6711b075153ba2ce26b0801c6df","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-28 15:01:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"206ed0f1801817edd8d09893f5b5dc1d2af737a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 15:02:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"366e63e2a2326b5bf85d2e3dd9795b0a7a078b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:58:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7aebd8927b94531b64a225a9d8e9943 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/988327a6ea88449c9d7e98d851be4b14 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/547fe75e203a475a86c2032e45a224b4 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b20e93c13c65420d97725c44138db9ed : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fe6cc2c194d4f65b0c70c482727c3d3 : SUCCESS in 10m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4de7515f72184379802c6e119c19af81 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4227afd8f3bb4ffd9535c3d8c10b4ab1 : SUCCESS in 9m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8d7cda6d026641f88fb8a3ead665a171 : SUCCESS in 8m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d069a903a624e059648070fa0fe825b : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1e99a9cff290461fb6d95e1e7949637d : SUCCESS in 6m 51s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/428b75ebdbb44ae383d99ac459d0965b : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/54bc48727a2f44deb46ad4a03a349855 : SUCCESS in 9m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9345979fb4d54cc5b1e653c8946723a9 : SUCCESS in 18m 27s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8277b34793894170ac64f5d682668331 : SUCCESS in 15m 08s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/7fb0c5171b5443169f85fde58d54f430 : SUCCESS in 7m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a84c518e3a2c41e2a4e799c255ac8a05 : SUCCESS in 28m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7025298442e44ff3ba6151e552311265 : SUCCESS in 14m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f06cbfcc247458c81ccf1c28e73c377 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eeb10ae6a83249618e6c3a6ea82a0309 : SUCCESS in 25m 13s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d4574666472642fbb2c8e5e4df272610 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/447b2603d73d48278ec8982cec634c31 : SUCCESS in 8m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4dfba557d9b74dc38bffac26b94cf116 : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88ce028ce9964a17ae1ae1903d9dc5e4 : SUCCESS in 43m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8749c7ec4d6a40a89b92f345c11d65c5 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b2d9b6315d91b5f27cbad2d3b734be5d47f1e2f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:58:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"116857de6c3335265deb226b789f6b6fc8069b83","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 19:01:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b463b6e14545e09154108319adf760 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b2abe813ba813efcbd950cbbca609c5b8e4bca3","revisions":{"2b2abe813ba813efcbd950cbbca609c5b8e4bca3":{"kind":"REWORK","_number":1,"created":"2021-04-27 18:39:46.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/41/788341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/41/788341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/788341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/788341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/41/788341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/41/788341/1"}}},"commit":{"parents":[{"commit":"752d36658db9c836e8cebb4bdd748f84c3165807","subject":"Merge \"update constraint for openstacksdk to new release 0.56.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/752d36658db9c836e8cebb4bdd748f84c3165807"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-27 18:39:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-27 18:39:44.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.6.0","message":"update constraint for oslo.config to new release 8.6.0\n\nmeta: version: 8.6.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Commit: Daniel Bengtsson \u003cdbengt@redhat.com\u003e\nmeta: release:Change-Id: I47e48b9f1a4f5b3f8f3525b0aed57c4db441c6f3\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I68d31601a337847ce646ac250521fb8809938014\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2abe813ba813efcbd950cbbca609c5b8e4bca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2abe813ba813efcbd950cbbca609c5b8e4bca3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
