)]}'
{"id":"openstack%2Frequirements~788409","triplet_id":"openstack%2Frequirements~master~Ie5578ea2b845760640aa509b82b80f66677b3c6f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"Ie5578ea2b845760640aa509b82b80f66677b3c6f","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-04-28 06:28:00.000000000","updated":"2021-05-01 20:48:56.000000000","submitted":"2021-05-01 20:47:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788409-openstack/requirements/constraints/noclob","meta_rev_id":"5cf8d0ea8d68a144aa546deb873c970ca70b82e2","_number":788409,"virtual_id_number":788409,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-01 20:47:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-01 19:31:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-01 19:31:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-28 06:42:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-28 09:10:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-01 17:56:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4abd9ea262a1e5bddb7ae84fb060c079ad43558c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-28 06:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9407c49103d0f9e4d1ecc207af0d0106df6df490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 06:42:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2385886d789e428c9aee6bbb16537798 : FAILURE in 9m 08s","accounts_in_message":[],"_revision_number":1},{"id":"7fee9724700ff91781bcd7629919223b4a3af897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 09:10:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a336a737a89e494d8b952cf200b71833 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81dcfcc902f0433b86c7a8ffa75c16ce : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9baacc5767f4fb1bc430bd6ace5363a : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35ce2f4f57b84130afd56977ad93aed7 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44b131993adc49ba9d8ae76d3ffb8c06 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbb8d4f1f4374f30b10ffb73296165db : SUCCESS in 9m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/256d510320214ac5b72255890fef5954 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da65a83d08444756aafee1c01893d6a9 : FAILURE in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72b2dff1b1a24f2c9a6c7c005e9e7261 : FAILURE in 7m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5b1e9273a65a4c92ba86a8489667f2ea : FAILURE in 9m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/599d3acfe3e94f08816132a779f63a2f : SUCCESS in 7m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3f03adcbffe048a8aedb1ca6d83bce65 : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8a3f3a155a2d4de1b9319c8f4a2d2690 : SUCCESS in 9m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/386536b32b9c4bbf81cd943e84661925 : FAILURE in 11m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0ccef09849d54095a3d50b55e52e32da : FAILURE in 12m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9926611697bc4505abb49756f7b038aa : FAILURE in 7m 06s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cfefa7500e85477ab535d8b37cd40a9c : FAILURE in 25m 45s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/cf1d31cf4e5c4efea00e5f1188c5f9d6 : FAILURE in 14m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a478b4c50e9e47539b8f15070865dfa6 : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cd16865f566a4de1a22835ca1a84c2bf : FAILURE in 20m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b027a3d60b8445d995e0c6fa0c38fe6 : FAILURE in 7m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ce97e4a1c91242459f05cb21788d0047 : SUCCESS in 6m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/36b7b5dd79d94860bb1b11f1c4986405 : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6d82ba8d36b4d8ebce1cb53a9797ec5 : FAILURE in 2h 12m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/955fd36cf85f4459a9ea3716a9ff9753 : TIMED_OUT in 2h 02m 56s","accounts_in_message":[],"_revision_number":1},{"id":"107cf2d12f2030d7a06464d0cdee938e5b92fc65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-29 06:22:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"762fb604c7f9599203b435620df6a5ae6263a33f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 06:39:33.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/483bae7c7a95493b8feec4c3ef682610 : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":2},{"id":"36267337730e99e37cfd036d0e1129939bdac085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 08:36:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b646e98b9d404d3fbce633606e58cf29 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af19e07c2f4148988d05c97e17094354 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e157ac34d02a461e9f2a33099c7285ec : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c910119f0da4c1694bf91a133e6966e : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4aeacd0771e4949b8d4fbc9fc0ce8ae : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d89698693ab4e2887f573c4e7d1ef7b : SUCCESS in 5m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8914e31036584047be2d7c6f64024682 : SUCCESS in 4m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5686a42cb45e44b3b306b4bcab382ecc : FAILURE in 5m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fa42f5addbcd46d6bc8bb59bb6b60e8f : FAILURE in 8m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e029ddc99e4e4f7087d4c49ea625f33d : FAILURE in 10m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/3c8d4ae2d3d04475b88eeaccbda721dc : SUCCESS in 7m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/8a8ecda5883b4d588e7d39a2d3cda96f : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48d2f6f9e7894106b32ed5ba3a47f7ef : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/50297fc34fef4ad2910843c330e70010 : FAILURE in 12m 18s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/63f361bfabf04d4a98a0ab5d7b877bcc : FAILURE in 11m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/cea4f92572ef453ba91966238cb9abef : FAILURE in 6m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/60fdfe4f6ae34d71945d440681a8ff87 : SUCCESS in 25m 06s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5cefe32bfed04b6ca456f48bfe31ed82 : FAILURE in 15m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/afa46c6b68e1461686182e76eb2e3880 : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9f5203112f224662a5bf3af8b112bd81 : FAILURE in 20m 00s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/bfe541b9cd1545efa92a6a29dd3c22d1 : FAILURE in 5m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/79b3bf48d4bf4296af80f937ee9658a0 : SUCCESS in 8m 56s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4ec877795eaf401288bd4c2d6a7bd717 : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8159bb051eb54d4587c68176dcf8e25f : FAILURE in 2h 08m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/01fffb24540b4ec7844bbe065abc79ec : TIMED_OUT in 2h 03m 37s","accounts_in_message":[],"_revision_number":2},{"id":"86f3c8772562256b736213a5618fc6436df1d2d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-30 06:24:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b5ef4cec7a2bbe069ac00a6ec1c2f9d921087a27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 06:41:30.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a36b74359fcc4a09a042219e46cb0a9e : FAILURE in 11m 48s","accounts_in_message":[],"_revision_number":3},{"id":"3c03ad9b5e68e84b1dc7039968157b2488def3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 07:54:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5350fd11d065443eaab178132e26955a : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5984551bbed49ef8d77e3de8d615d19 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d7da32d5bd34d25a0455cf22751a6dc : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae19b999a97249d986f4058f3e52fa23 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dda4dd972bcf438782eeb1c3828c59ba : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95e8f4ce176d427bbd702438e4839192 : SUCCESS in 6m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36dddf4de057437da9a1f20c99dc9466 : SUCCESS in 4m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2efde458d6a24783aaf5346a25985333 : FAILURE in 6m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85c565140293496982a840348b6c5417 : FAILURE in 5m 54s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d18698950eab430fb9b990d1c87c4df8 : FAILURE in 10m 20s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2f9d76b52151405dbaff1b1d3d8122a4 : SUCCESS in 6m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c68cfdef851e48279c438c15ce72d0e0 : SUCCESS in 10m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96c1dbb2ab8946a19e5d4ff576c2ef77 : SUCCESS in 9m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/beaa84bac5e342798976286e4ca062c1 : SUCCESS in 11m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/ca247e54025745358c0ccc75b0e3be58 : SUCCESS in 16m 09s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/095440a4a09c4dc6890a435d2376a667 : FAILURE in 6m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/51946e83897d49989ebccdaba9069f69 : SUCCESS in 30m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/1596314383684098b916dac36ee89c58 : FAILURE in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a5a56dd6efc4b4ab46f4a674e9fed36 : SUCCESS in 5m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4307efbd1f1d4bf6842dd0b1c89a0957 : FAILURE in 21m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f65e8ae2692742b787ccfedcf276da75 : FAILURE in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a446747f9dc41c695a6a55b0f0e9e11 : SUCCESS in 10m 12s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b0c00212a8934ae98d259ae96036f116 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c047a219e9144452944cf1fb81177a02 : FAILURE in 47m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8552893ccc04c2baf1a11ad45d4855e : FAILURE in 1h 01m 56s","accounts_in_message":[],"_revision_number":3},{"id":"8d4631a086c793598ccc089ab91644e049a3019d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-01 06:33:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"549ef536e8ea1fb4948c1ee6605bc9eb21c494c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 06:49:11.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/58c08a86fa2e4379986c76e9735706d5 : FAILURE in 11m 47s","accounts_in_message":[],"_revision_number":4},{"id":"c82eeb232c18a86e95a566d12c057e94f23f909c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 07:42:12.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef4238c3ec104bef90e59bfb936db034 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ebd80387e8448eb67898fa2d118a55 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fcc0465162794aa4b101666cc22b3a08 : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6015b63557124eb19e2a4377d0ead8df : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99937f1a840b42099c61991c18b16487 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8153e785be614a4aa8857463b7dc879b : SUCCESS in 5m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2549fc0414b48429a6a09fc2e704248 : SUCCESS in 5m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c9a9e4644d9a435c80c214d9e7307f1b : FAILURE in 6m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70ca57d78f89441e91a2d0f7e96a398e : FAILURE in 8m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e2d61c4fbfd14e65b08d7da690adf447 : FAILURE in 8m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/63e423597f824cc8a4c1c6c3efcaed4d : SUCCESS in 6m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/92d6bd576c654f81a50a2fe2b664643d : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ffc3e998f17249068b8d0efed77be42a : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a239e16b21eb49ebbe55c2624535714b : SUCCESS in 9m 56s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/07f1df930b69471080f70d068e44f68e : SUCCESS in 12m 07s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a85f640cb6d04b2cb5cb086ba3615041 : FAILURE in 5m 25s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/96c0862d3ff840f7a6eddc29978094bf : SUCCESS in 25m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/394b1bb983784ecf8eae0cf1ae1fc957 : FAILURE in 15m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6fdc3b69337a4fd2b40243d66c0e4dab : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56ed875cf1b744988dfcad6f00926b8d : FAILURE in 19m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c816cd3d4124466fb160dc813c0cd43d : FAILURE in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/80252e6bd53b417a861c1f9dad230614 : SUCCESS in 11m 28s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/3cf0da8174a04f2c83aaf44e1f51f095 : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/799b3b3c8d2d4ee89947bacdf024054b : FAILURE in 50m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c2a946d2a421450bb145c19222162831 : FAILURE in 59m 20s","accounts_in_message":[],"_revision_number":4},{"id":"e097e7b9818022645248c539680f38623ac72ebd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-01 17:56:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a7d413aec0fc6fae62efb8471bdfb1a964826eb2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-01 17:56:11.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":5},{"id":"59db2fdcaf45fc3312f250bcd34f58c82bd042d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 18:15:08.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a852f1a59708417eb8c951db123ae2a9 : FAILURE in 17m 16s","accounts_in_message":[],"_revision_number":5},{"id":"ea6a9f7f06b6a56930a524d5c8d9ff56b8499fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 19:16:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6d67ea6a8584bcc90eee33170d8c7d8 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/554952756cbf421e8cb3b3c1068e9c02 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16c1116469d14dc5ba90f1a5034c4ff8 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b47e1eef75640caabefa5cdb1fd4252 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b72a99f9edd24e74bbad0ac67352e80e : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4c44c79b61444e8aae5db4826c2004f : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3347f50e115f4287a73095d50312d4c3 : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bdb42c6a9cda4295ac8423014648538b : SUCCESS in 10m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/76238b7d1e1c4159b69ea1cdcfe589f1 : SUCCESS in 9m 05s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/70879839062f41e3a6f63845a1391c21 : SUCCESS in 9m 28s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a88c17531e54474cac3026b17e481de0 : SUCCESS in 5m 30s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/70d3eeef2be648dbaef0f3239cff528c : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/97b47aec4e764dbaaba95b7f2cd11cfd : SUCCESS in 7m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77059acccf81441aa965a0889fdd17ff : SUCCESS in 10m 42s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e10ba00784554aec96a8ba600b2706e6 : SUCCESS in 13m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/faeb3fb275a74f0aa2bd513ec76c119a : SUCCESS in 4m 46s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/602e117c18f14845a19e701010c5d658 : SUCCESS in 32m 54s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4bf7c32c54ee4e5fbd091000e88b6016 : SUCCESS in 12m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a27039f7087043a0b9af04ff75b130c4 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fa6e9a725fd4355b8f1e68b4f0bc863 : SUCCESS in 13m 35s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2016ef002ee14451b4e3fc14aaaeb308 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/133cc77c62e246bc8f0328647deedac1 : SUCCESS in 6m 30s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/fb474a63a6e8477f9f8a0cdfdba55eeb : SUCCESS in 12m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91d9aa86d3e34e27ba5da9f49af05109 : SUCCESS in 54m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fcc987b19e9c4dc58ad6680c0dbe8422 : SUCCESS in 1h 15m 59s","accounts_in_message":[],"_revision_number":5},{"id":"c6c6e43a6de7808c3b9addffa9fa260d06d94dd7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-01 19:31:21.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c39299a627cdb5a06c9a87c69cb63c79441b326e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 19:31:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"18b4f7b61fcb4ea9e565ccdb595463ad2a04279c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 20:47:49.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd7a0f74b3ec4a14b7a78d01f115b25c : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596779a3b19745a982317fdbb6f9efb1 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc6f67ed558b4973adfb3c5fe653d28b : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c968b9fd751941b4a64e93a07c00224a : SUCCESS in 4m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c326271417b4fac9deb1108f4b131ba : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/877c0f3bab6447d198bea8491a56e2c3 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/38b1c1e0e3104d14b4f577269adb7570 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1de9add535b2470c80c5d9196f366ce8 : SUCCESS in 11m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4d223ec3e4ef4a8ca63d0d9c18747a5f : SUCCESS in 7m 51s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/89cba40e9a3d465b8ff35a6a9c8a4e9f : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2644bbf2514148b5b36242d570d95426 : SUCCESS in 8m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3f3000f44104963bd1a81dd76afd933 : SUCCESS in 6m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e267fc02eead4b63886197e6f531cc91 : SUCCESS in 10m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70ccd551ed1849f3b560677849bcd71a : SUCCESS in 12m 09s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/b3c378e5b900452d94dd75f2d56d5a9e : SUCCESS in 4m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7a4c7a9d0464424baea7bc447ddfcac5 : SUCCESS in 22m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fedbb5353a63423781478b7f8ec8c085 : SUCCESS in 10m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1aafa7ddfb4e4b67a150681444f2e063 : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/675ca6b370f8444f9def98c92707cb1e : SUCCESS in 25m 21s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0defebb1388b4f44a904c960a885ab61 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5493ec2fc9834fbf83388e49b6fdbe95 : SUCCESS in 6m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/91292707c52244dfa678bd179b151638 : SUCCESS in 9m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec2eb05ae5eb4cb39f35c75bd0dada5f : SUCCESS in 43m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f51619c349cb49c7b837522ed6dce46f : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":5},{"id":"43b5766dceaf09b70e94d0f095f052e8d4288759","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 20:47:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5cf8d0ea8d68a144aa546deb873c970ca70b82e2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 20:48:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd77db1a1f74436ade364c2ea8145a6 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3e29470ec4903e4485127060c5bc04c359827552","revisions":{"66e0b1e3165ac7213e66354da42fdc8e0915925a":{"kind":"REWORK","_number":1,"created":"2021-04-28 06:28:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/788409/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/788409/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/788409/1"}}},"commit":{"parents":[{"commit":"752d36658db9c836e8cebb4bdd748f84c3165807","subject":"Merge \"update constraint for openstacksdk to new release 0.56.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/752d36658db9c836e8cebb4bdd748f84c3165807"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-28 06:27:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-28 06:27:57.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie5578ea2b845760640aa509b82b80f66677b3c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66e0b1e3165ac7213e66354da42fdc8e0915925a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/66e0b1e3165ac7213e66354da42fdc8e0915925a"}]},"branch":"refs/heads/master"},"3fbdc2a5bfd26ed96e6386be57a4c6e85a78d6e6":{"kind":"REWORK","_number":2,"created":"2021-04-29 06:22:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/788409/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/788409/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/788409/2"}}},"commit":{"parents":[{"commit":"2b2abe813ba813efcbd950cbbca609c5b8e4bca3","subject":"update constraint for oslo.config to new release 8.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2abe813ba813efcbd950cbbca609c5b8e4bca3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-29 06:22:38.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-29 06:22:38.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie5578ea2b845760640aa509b82b80f66677b3c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fbdc2a5bfd26ed96e6386be57a4c6e85a78d6e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3fbdc2a5bfd26ed96e6386be57a4c6e85a78d6e6"}]},"branch":"refs/heads/master"},"a526a808e0d2ad46bc2f436bfa201ab934197791":{"kind":"REWORK","_number":3,"created":"2021-04-30 06:24:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/788409/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/788409/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/788409/3"}}},"commit":{"parents":[{"commit":"670a69b3a0c14b0ea764de2ed484a42c17d7edeb","subject":"Merge \"update constraint for validations-libs to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/670a69b3a0c14b0ea764de2ed484a42c17d7edeb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-30 06:24:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-30 06:24:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie5578ea2b845760640aa509b82b80f66677b3c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a526a808e0d2ad46bc2f436bfa201ab934197791"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a526a808e0d2ad46bc2f436bfa201ab934197791"}]},"branch":"refs/heads/master"},"43fa482e8fda5e2c43d4de73010602e520cc7ee4":{"kind":"REWORK","_number":4,"created":"2021-05-01 06:33:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/788409/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/788409/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/788409/4"}}},"commit":{"parents":[{"commit":"670a69b3a0c14b0ea764de2ed484a42c17d7edeb","subject":"Merge \"update constraint for validations-libs to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/670a69b3a0c14b0ea764de2ed484a42c17d7edeb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-01 06:33:25.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-01 06:33:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie5578ea2b845760640aa509b82b80f66677b3c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43fa482e8fda5e2c43d4de73010602e520cc7ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43fa482e8fda5e2c43d4de73010602e520cc7ee4"}]},"branch":"refs/heads/master"},"3e29470ec4903e4485127060c5bc04c359827552":{"kind":"REWORK","_number":5,"created":"2021-05-01 17:56:10.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/09/788409/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/788409/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/788409/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/788409/5"}}},"commit":{"parents":[{"commit":"670a69b3a0c14b0ea764de2ed484a42c17d7edeb","subject":"Merge \"update constraint for validations-libs to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/670a69b3a0c14b0ea764de2ed484a42c17d7edeb"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-01 06:33:25.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-01 17:56:01.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie5578ea2b845760640aa509b82b80f66677b3c6f\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e29470ec4903e4485127060c5bc04c359827552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e29470ec4903e4485127060c5bc04c359827552"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
