)]}'
{"id":"openstack%2Frequirements~788800","triplet_id":"openstack%2Frequirements~master~I02e9175eabda0aa3ac57ed0f2813cdb5aede2013","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I02e9175eabda0aa3ac57ed0f2813cdb5aede2013","subject":"update constraint for ovsdbapp to new release 1.10.0","status":"MERGED","created":"2021-04-29 16:43:39.000000000","updated":"2021-04-30 00:34:48.000000000","submitted":"2021-04-30 00:33:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788800-new-release","meta_rev_id":"be798ae2985a1de31e4d2ded90a5e5e3b1a8ced4","_number":788800,"virtual_id_number":788800,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-30 00:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-04-29 23:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-04-29 23:04:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-29 17:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-29 18:15:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 23:04:45.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"18bd803fba2838f6f79d8eb85610bb9e332f488b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-04-29 16:43:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e3b0b6411663e5a26cb74bcae91931ffc1e67bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 17:03:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f58740752b3f4779ac2804e8f9cb64dc : FAILURE in 17m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cf601d4ffb51a56e925de6fc1183ae3ed1e720f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 18:15:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e17cfeb05c4afab4b2d12b302a92f5 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1b83359a7e41aca00935b15c96299a : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34cd9023499e43bd890add4cc56f579b : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82c525a524dd485bafc1d35bb2aecde3 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/577793aa5848442982a580ebd6ff0566 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/27e6c0ef66c844e99e177bd8f62459bd : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/62ad42e78762418d8ecc83bc6def30d8 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8f329bc5a2b14629b2552cdd8c49f3de : SUCCESS in 10m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/92df55756b9e4f4da30a2ef91cd69a02 : SUCCESS in 10m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b7933eeffd9d414e81a33b089e3a36b4 : SUCCESS in 10m 11s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/67a8e99279ab4f9fb38795d2431a6223 : SUCCESS in 6m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/96ada4482e064848bc9c697a984f768d : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/82b894e07b9e4fa98a9461acd8411ad2 : SUCCESS in 7m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc6e036496724b40a588167eb32da3b6 : SUCCESS in 11m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/74e3b66207214624a267c597a708c2a7 : SUCCESS in 14m 11s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fa348667278c4877a58ae9c8f002851e : SUCCESS in 6m 02s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7185d523b3b74285afd88c56e2997e5a : SUCCESS in 36m 41s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9c5b83db464049c7b37b8294d5fbf51d : SUCCESS in 14m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/030bed4ce7cc42b5a912751ebce978bf : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ab7e5bbc4e58445a960818703fd1e679 : SUCCESS in 22m 12s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/daa67584845b4aa3a7a63d8a9690aafe : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/18c92a73aaa44c7c9f38e1319747a3ac : SUCCESS in 7m 16s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/71f59e3fb9b647e48b009a3cbbb3bb01 : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c56d922dbf640bfa46871c1bd257d90 : SUCCESS in 56m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/554ab10c27054f9daa152883e6b06c95 : SUCCESS in 1h 26m 17s","accounts_in_message":[],"_revision_number":1},{"id":"b5e64844f5ef9bd19314c4878f518e4c3baaa848","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-04-29 23:04:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fd1808d70f3fe5cb4101b517ebeffcb9509b10f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 23:05:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5e02593c0b00bf4fc87109b8455fd3746177270","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 00:33:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba78d410c85749febdef94bd246508f3 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccf9b7e778044a0fb57b23ab83c6166b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a05cd03a39da4366843e0a4912b5786d : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86af48c57e854dfab0d3e99651129abc : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8c0679de0064203baea6593f3c553db : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1766314741f4327bd7c003117aaf778 : SUCCESS in 4m 05s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/643fae7699fa454d8ad0b7fcd8526ded : SUCCESS in 10m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b91d5163f6234edab9746a947fc7c257 : SUCCESS in 8m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/148502edb1ab4286b8b5cede92e606e7 : SUCCESS in 9m 48s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4b5dc36a377f47998421160653c1a5f8 : SUCCESS in 5m 52s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/766b37c1982f4d62b79da131237c379c : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5096435fd3da42b2a15a5d11c846ac89 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b25e59a8f5cc470caef9d271a0ba2a58 : SUCCESS in 9m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0c2c626edb9b4d99a9fd3fa4dbb328fb : SUCCESS in 14m 12s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5b498376cdbb41619a69536e23dd42ca : SUCCESS in 5m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3d9f3b7d658741ab90fe3f72e3fd6ea9 : SUCCESS in 22m 53s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/326e3e764f924ac2afa3bd8e4d44d75e : SUCCESS in 11m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9869efd09b04ff080010b48cacc73d6 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2b0f68b9e9144b74afaa16d56a9b0481 : SUCCESS in 15m 41s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4e38089c80264863a7f4a0c4ac18a2a4 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67d1955948d94cfc9c6a077b560b9b19 : SUCCESS in 7m 13s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4a1151209e3b43828e78018e605da7d5 : SUCCESS in 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d01ac8f053e44db7b599792565ad4e48 : SUCCESS in 48m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e64f2e42d8b41bf90c801fdf4172ec4 : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a8fe0ebfa223435e5100aabddd3fc65b7680461e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 00:33:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"be798ae2985a1de31e4d2ded90a5e5e3b1a8ced4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 00:34:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a2a23065d74118a3592ff422f5b2f0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2c0a311ec58f54c4d8ae10acd0f2fff1000a08f8","revisions":{"2c0a311ec58f54c4d8ae10acd0f2fff1000a08f8":{"kind":"REWORK","_number":1,"created":"2021-04-29 16:43:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/788800/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/788800/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/788800/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/788800/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/788800/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/788800/1"}}},"commit":{"parents":[{"commit":"2b2abe813ba813efcbd950cbbca609c5b8e4bca3","subject":"update constraint for oslo.config to new release 8.6.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2b2abe813ba813efcbd950cbbca609c5b8e4bca3"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-29 16:43:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-04-29 16:43:36.000000000","tz":0},"subject":"update constraint for ovsdbapp to new release 1.10.0","message":"update constraint for ovsdbapp to new release 1.10.0\n\nmeta: version: 1.10.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ibb5eac49747fd8cd71a819083fd33e72c70db1ab\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I02e9175eabda0aa3ac57ed0f2813cdb5aede2013\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c0a311ec58f54c4d8ae10acd0f2fff1000a08f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2c0a311ec58f54c4d8ae10acd0f2fff1000a08f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
