)]}'
{"id":"openstack%2Frequirements~788912","triplet_id":"openstack%2Frequirements~master~I021c32fb063417f20899c93a0ee3effccdf496f6","project":"openstack/requirements","branch":"master","topic":"toml","hashtags":[],"change_id":"I021c32fb063417f20899c93a0ee3effccdf496f6","subject":"Update upper constraints toml\u003d\u003d\u003d0.10.2","status":"ABANDONED","created":"2021-04-30 09:49:19.000000000","updated":"2021-05-02 11:10:06.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"224f7a335506fa0fc9e3e67ed7d559c045569d16","_number":788912,"virtual_id_number":788912,"owner":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2021-05-02 11:10:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 10:02:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-04-30 11:04:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 11:49:36.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-05-01 22:14:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8a7d33d0101f86afaa46373e52eb93d7efdc7c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-04-30 09:49:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"033638e50f75c852a3ae7443be828d7fca7e6594","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 10:02:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/51e386b47ce4460aa4c63731887fae20 : FAILURE in 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"efb9afdb915063861197d5655928f9fb8c393205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 11:04:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0655a1206b5541b099501c820782d2cb : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4363690c554b4f71a11c18ed5b6841b0 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7570716792f46d59c1b6ec54e92d8c4 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63836ebcc6374313b90dc79bdff88c22 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93ca4269a2cf45f08f8c165efd5152b7 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3431aa8da2404ed5a773ae6522eca99d : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e8e97d29247f4a08a1c3183f7bbc79db : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/191782aaaddb4074b8284a5ee492a3be : FAILURE in 5m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64400c13aefa44428b4fee934376a9ce : FAILURE in 5m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/23caff002b034791aa58b069fdb8e665 : SUCCESS in 9m 05s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d97e0512707742178b127e206dff17c0 : SUCCESS in 5m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/176690547c4348eba79e072dec8304a1 : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08bcfa7e39f54665b91b3ace0713825f : SUCCESS in 7m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ee0b6dd1d4ab4c31bca37ef5b4990f3e : SUCCESS in 10m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/21e01e9d3934499181a585be9d9a104f : SUCCESS in 17m 49s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/051969159adb46d2814dfde54121f4b5 : SUCCESS in 4m 37s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/11b6468c4f8242e091fdb3a9b9b54fea : SUCCESS in 26m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/14dfce2e6f604d2ba146c92eca8f23ff : SUCCESS in 13m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/727e629bf0d74ef38372aaca14a9ddfa : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/95e1040c8458409fa8c00664260082d8 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a3b283ac57584d289962c7a4d494c17b : SUCCESS in 5m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3bdaaccfa6c04714b4591235dc31c218 : SUCCESS in 7m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/849c706247004422beb104babb0757db : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f6d265781a243dab9ddf0bf2106e353 : SUCCESS in 49m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87788721f7e94404be50b9bc37ef6ff1 : SUCCESS in 1h 13m 09s","accounts_in_message":[],"_revision_number":1},{"id":"68e054827a0e1e635eca31223a89611b74f34d21","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-04-30 11:49:36.000000000","message":"Patch Set 1:\n\nLooks like there are other dependencies that will need to be updated as well in order to bump toml:\n\n```\nThe conflict is caused by:\n     The user requested toml\u003d\u003d\u003d0.10.2\n     confspirator 0.2.2 depends on toml\u003d\u003d0.10.0\n```","accounts_in_message":[],"_revision_number":1},{"id":"bf31ca296d870a728e7494478b8487415e2bd9f9","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-04-30 13:57:55.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Looks like there are other dependencies that will need to be updated as well in order to bump toml:\n\u003e \n\u003e ```\n\u003e The conflict is caused by:\n\u003e      The user requested toml\u003d\u003d\u003d0.10.2\n\u003e      confspirator 0.2.2 depends on toml\u003d\u003d0.10.0\n\u003e ```\n\nIndeed. There is a fix for confspirator that is waiting for review [1]\n\n[1] https://gitlab.com/catalyst-cloud/confspirator/-/merge_requests/2","accounts_in_message":[],"_revision_number":1},{"id":"b4237aee60fa8620b2beb157a373e012eadf40d1","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-05-01 07:59:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"54ffae9854f566348183e93067bc5fe6da6ef5c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 08:12:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a2573de99bab4f21bc842452085169b2 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":1},{"id":"45739ea6e84e59c4ddcd1eb48eb84936a3dfa84a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 09:19:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2191fad715a54997b20f818895616219 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0861d7ec884b4261917ccd03530e3276 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ba83b7842794b969804c237b7e4fbe2 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9395b790064e48a3b0c4c481d0b28330 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/509ca9195801421b96299c6892156994 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4f139db81b914c42a0382d143959aa8e : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98b2956a41d34a369d8b957a4bb4e932 : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e6888a4b32384d10966d5d03f2fb1261 : FAILURE in 8m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0def011df44e4a998c3394cd062ec2c6 : FAILURE in 5m 42s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6d72c13db449489dbb9e3da4e07217fb : SUCCESS in 12m 29s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d6f109670a6c42c093f4c174ca656f0a : SUCCESS in 7m 27s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2db68a2cdcda4d2bbe0349b823b9921b : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0af63c0f19c64945bd3d4c147351987f : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/60881c30ed354b489022bccf66c05d91 : SUCCESS in 11m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/59259588946f400d8f0b159b43408ff6 : SUCCESS in 17m 20s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/73a68b82360c45b6a10c79064312d689 : SUCCESS in 7m 52s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/56f24fe1bed0416c9e245b4addf40bbf : SUCCESS in 27m 09s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aa80445b426a4b59a7e2a7ed9cd26027 : SUCCESS in 16m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c619b71236814256abce56a900d532ac : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6d852ee4abe9425bab92bc110c040823 : SUCCESS in 15m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/98fd6cc1f20f43418917decf36620e77 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f4aca0dfb06946e890ccb9e3763773f4 : SUCCESS in 5m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/efe37109b4d743578726a54f79acefc6 : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83d84cbb74b44c78964b27514504cb39 : SUCCESS in 57m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da00acded636473db6417d92b5abbafc : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":1},{"id":"2a56d54bdbe2f9baa3069114c9edbe55e67b2572","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-05-01 22:02:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f756d62587bbed955b9e2809b04d65f69e107b6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-01 22:14:43.000000000","message":"Patch Set 1: Code-Review-2\n\n0.10.2 is already set","accounts_in_message":[],"_revision_number":1},{"id":"de502ac0ba67f954f639f45904e9dc9bc0ad6577","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 22:19:47.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a3b130aacf9f4d59b41081aba4c12375 : FAILURE in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a4ec09577d0015f2cd63932d08109b14cec9bc65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-01 23:16:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7734394791415096eb330bcc95c472 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fba9da604c748edb9db7987d0c00d1d : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a726bfb9b39946f8922790b660918c0c : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1c07b30101046639201121a50903d8b : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/717d2705388a443cb0c4d8818c6452d0 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/77b439997407473c9b2bedda440f59ad : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e41b9ae54524f5fa1931225e37cbdfc : SUCCESS in 4m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8aa9fb913d5a49938c0e561516507426 : SUCCESS in 9m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84ddbcb8edeb4573812115f05c631df7 : SUCCESS in 8m 45s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7df19cbb1e76465cad51d47b8b3f86fa : SUCCESS in 9m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2c0e1fb9c54548a995a4e81bdc59f3b7 : SUCCESS in 5m 56s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5366156760b7499bad1f4c2cdf92cdcf : SUCCESS in 10m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7e3e988049c240c4aade98f33c90ec6b : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d0eca8e4f73343d2b2beb1e5363d5dad : SUCCESS in 9m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9f7365805aa041b88ee298ddaec62e5e : SUCCESS in 18m 04s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/56e77baed89346a1a50f10074c245bb9 : SUCCESS in 5m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/41e69c1c472044989629d8ae0a1369ff : SUCCESS in 29m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/3e81bfe592f24665a0b0c80b6b1581e5 : SUCCESS in 11m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ee5903123ce44ed290b41296f6fbdcd7 : SUCCESS in 5m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/091d09b368f44086959b744a16f30c87 : SUCCESS in 15m 43s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0f0338c3731d45c0bfc58aa88738fedf : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/43cd3da1e327471da58854b70914e332 : SUCCESS in 7m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a2602fb5601d4c61b899d8c89a918e46 : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55ff99d447cd4137a6ac91cacb7deb26 : SUCCESS in 46m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a97ca12254c84b318d2b24b95c1c8ae5 : SUCCESS in 1h 13m 02s","accounts_in_message":[],"_revision_number":1},{"id":"8fe9eb05880b5b5d4fc184c40446321af6edec7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-05-02 11:09:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"706c651f348dab30a6ac579a5adfa11c44256a3d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"date":"2021-05-02 11:10:06.000000000","message":"Abandoned\n\nThis has already been fixed by another patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3ceae5d6e62fcafc807b1d6f909590f91ae4ebc7","revisions":{"b3d80eee1c4fd2552e54e4939a62e772c6fee8e8":{"kind":"REWORK","_number":1,"created":"2021-04-30 09:49:19.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/12/788912/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/788912/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/788912/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/788912/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/788912/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/788912/1"}}},"commit":{"parents":[{"commit":"670a69b3a0c14b0ea764de2ed484a42c17d7edeb","subject":"Merge \"update constraint for validations-libs to new release 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/670a69b3a0c14b0ea764de2ed484a42c17d7edeb"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-04-30 09:49:10.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-04-30 09:49:10.000000000","tz":120},"subject":"Update upper constraints toml\u003d\u003d\u003d0.10.2","message":"Update upper constraints toml\u003d\u003d\u003d0.10.2\n\nChange-Id: I021c32fb063417f20899c93a0ee3effccdf496f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3d80eee1c4fd2552e54e4939a62e772c6fee8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b3d80eee1c4fd2552e54e4939a62e772c6fee8e8"}]},"branch":"refs/heads/master"},"3ceae5d6e62fcafc807b1d6f909590f91ae4ebc7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-02 11:09:28.000000000","uploader":{"_account_id":27329,"name":"Federico Ressi","email":"fressi@redhat.com","username":"fressi_redhat"},"ref":"refs/changes/12/788912/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/12/788912/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/788912/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/788912/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/12/788912/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/12/788912/2"}}},"commit":{"parents":[{"commit":"3e29470ec4903e4485127060c5bc04c359827552","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3e29470ec4903e4485127060c5bc04c359827552"}]}],"author":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-04-30 09:49:10.000000000","tz":120},"committer":{"name":"Federico Ressi","email":"fressi@redhat.com","date":"2021-05-02 11:09:28.000000000","tz":0},"subject":"Update upper constraints toml\u003d\u003d\u003d0.10.2","message":"Update upper constraints toml\u003d\u003d\u003d0.10.2\n\nChange-Id: I021c32fb063417f20899c93a0ee3effccdf496f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ceae5d6e62fcafc807b1d6f909590f91ae4ebc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3ceae5d6e62fcafc807b1d6f909590f91ae4ebc7"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
