)]}'
{"id":"openstack%2Frequirements~789906","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ieff5bd70c1de8774be92ba884047e2f00445e94a","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ieff5bd70c1de8774be92ba884047e2f00445e94a","subject":"update constraint for os-net-config to new release 14.1.0","status":"MERGED","created":"2021-05-05 15:46:40.000000000","updated":"2021-05-06 04:42:17.000000000","submitted":"2021-05-06 04:42:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789906-new-release","meta_rev_id":"eb26267b60aecadec4592629ff47646fd70f1c32","_number":789906,"virtual_id_number":789906,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 04:42:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-05-05 15:49:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-05-05 23:34:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-05-05 23:34:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 15:49:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-05-05 16:08:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 17:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-05 23:34:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5e34fd9c9989725f5dd6331dbdc45fff295b6f58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-05 15:46:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b68f4b45739e85c66f921d11bcb2592a52cd0ce","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-05 15:49:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b67cb8ec28b2b42d640b39ba125fae46c758a731","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:08:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5c9192605f184fd7a1a9862d85e53403 : FAILURE in 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"48459f7e2557164080384c510762811406ad46f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:28:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094c4055d1714436ad1f74e88b24d091 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6b4d4deddea489f8c8f9dc140fcb907 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77cacd4e46c94562a30c5d7cb136a661 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddef77cfd8d248d9a630770f5b7e99a1 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/503bcd3f885040009b69d407ce7daebc : SUCCESS in 4m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e021837621324c999e1c08cd7701a26d : SUCCESS in 5m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/aa879638ed9b4bcebc75d444c4e0a2d6 : SUCCESS in 10m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64046a31b32d4ecb90d0006dcd96351d : SUCCESS in 8m 49s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8d10acbae968455ea09fa1fc82b278e8 : SUCCESS in 9m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/81377aee5e9947fc9d8d8ebeb2af6b9e : SUCCESS in 5m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/17b1535a9b4f4e1191abf21ba17da32f : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e2d02a1fae0844fa8ae444bbf5184cc6 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d841a3ab443b41edaea1e76ce950d8cb : SUCCESS in 9m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/70719548c4e54f2b9672e324a2eff18f : SUCCESS in 12m 01s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/88476b70623245e7966833bdfa7300a0 : SUCCESS in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8e282d179ac74088840f7b7a0a91e672 : SUCCESS in 24m 56s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/5840e5af35354e2599a7d93ebbf2b0d9 : SUCCESS in 11m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29df053a648649ee822937ade166a64c : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dc98489cb2f344479edff67e662af394 : SUCCESS in 15m 27s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/3e9cf447859e459bb4630cf246008966 : SUCCESS in 5m 28s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/730818088e0f44c7a630a9d888665a60 : SUCCESS in 6m 26s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2a851e6da1e14883b6ab7b91ce54c1c8 : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0301ecf5495041c08900178eed9214ef : SUCCESS in 44m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9f11294267247e1b2fb44365ed52c70 : SUCCESS in 1h 37m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e4525c850e2897024434caa59a02fa110414aca2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-05 23:34:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c7731237527be78314f5d3be6d4a5b9d4fd2985d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 23:35:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdaa385d8bf39768f789896893eee0876f857dc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 04:42:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62178fe5e6444844a1de4a4534b7e162 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5504e6d57bee4eca87d3c7f51b81e714 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b07879113944943b870893a72e4c8bc : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c65608f0e86946438a8b150e21fc03c7 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/83b23c9d97ff4d26a389678dca9fb5c2 : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b8833fdc9392419bb42ef6dab1a433ed : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db414d1878f64310be71982fcc59332c : SUCCESS in 8m 41s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ec6426e6cb234135bc22a20eb5a1b796 : SUCCESS in 9m 38s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6451d277b972431aab1da0562eb0d65d : SUCCESS in 6m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/506fd60d01454502a559d7e097beb138 : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c0de46cde7749ed9d9986c6c8b59a9c : SUCCESS in 7m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98b5d77da1c44c7baa6533d96946453a : SUCCESS in 11m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/9085cc48a30c47d5ab4c06ce55f6108b : SUCCESS in 13m 15s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/2a1eed5eec42445aaeb216f9e4aa71b2 : SUCCESS in 5m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3a37d2b75fb94f13b9dcbd88c14d4f71 : SUCCESS in 27m 55s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/25c90e4f3aec4e2fba060a7777128f67 : SUCCESS in 20m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c3c81e1cf2246228e206b4fc6a4bbf0 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/070c16fe45a847b7891659d59680be36 : SUCCESS in 19m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/987aaad2a1004abf8988f543500e4524 : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/202d7ca13bff430a98a76cb50a96833e : SUCCESS in 7m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/919282af0f4a4d3eb8dadc74e2c42822 : SUCCESS in 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62a52bafc559473cb2de72a2674c2530 : SUCCESS in 54m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c288eec3e219449da5a69d203f7f3c9a : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"eb26267b60aecadec4592629ff47646fd70f1c32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 04:42:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0babaafb67fb291f9684ef2a38fc2bf8cb1bcaea","revisions":{"0babaafb67fb291f9684ef2a38fc2bf8cb1bcaea":{"kind":"REWORK","_number":1,"created":"2021-05-05 15:46:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/06/789906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/06/789906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/789906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/789906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/06/789906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/06/789906/1"}}},"commit":{"parents":[{"commit":"a34c9d00c90dbfe5dfc30d8b61db69ee16c532da","subject":"Remove Xenial wheel build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a34c9d00c90dbfe5dfc30d8b61db69ee16c532da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-05 15:46:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-05 15:46:35.000000000","tz":0},"subject":"update constraint for os-net-config to new release 14.1.0","message":"update constraint for os-net-config to new release 14.1.0\n\nmeta: version: 14.1.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ia064530696cb5c47b009967f665900c2dd06d44e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ieff5bd70c1de8774be92ba884047e2f00445e94a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0babaafb67fb291f9684ef2a38fc2bf8cb1bcaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0babaafb67fb291f9684ef2a38fc2bf8cb1bcaea"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
