)]}'
{"id":"openstack%2Frequirements~789907","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I1d4cf4e9d00fa35e7a324f927c929e8720bd13c3","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I1d4cf4e9d00fa35e7a324f927c929e8720bd13c3","subject":"update constraint for tripleo-common to new release 15.1.0","status":"MERGED","created":"2021-05-05 15:50:54.000000000","updated":"2021-05-06 04:42:33.000000000","submitted":"2021-05-06 04:42:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789907-new-release","meta_rev_id":"dc29679beca1d903ffd3685ef09bf1e89fecb7f5","_number":789907,"virtual_id_number":789907,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-06 04:42:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-05-05 15:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-05-05 23:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-05-05 23:34:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 15:51:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-05-05 16:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-05 17:39:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-05 23:34:17.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"accb1282b22d47a6243c80d3296d90ed7df08b57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-05 15:50:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcfcdce056ea1102f251a0582b09adc46ece3d19","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-05 15:51:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"007d8013da4c4b7ca7d002b2c987b46e69cef399","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:14:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/478a765dfefd44bbab9754ffb1586d16 : FAILURE in 21m 03s","accounts_in_message":[],"_revision_number":1},{"id":"300e14d8f1b67c928e90bd9e5663167c10fddc59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 17:39:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dbd5d4d9a7448e09f290a8271833178 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0055a9c3d9ae491dbc837d4f8a1025f5 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a46ca4a7fe1499bae8813924386b1d9 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69d60f6b140b4312aefbdee32b7eab5c : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5212e19439544976acf7264fc2646369 : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7265e57e904347f7af9d02a6246a0168 : SUCCESS in 4m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4d11f6782a6e4817ada918fde58ef8ec : SUCCESS in 10m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b7ca77ab5ea240ba8f84da54ae4a549d : SUCCESS in 9m 14s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/7822f9f19bf74b198a1bbfa656d9e016 : SUCCESS in 11m 36s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/59e5d491d8f642ceb44c648863ac55f8 : SUCCESS in 6m 06s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/31c40dd17f19434a96f5bdb73e335966 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3670790d1c5743d6915c8a52918f29ba : SUCCESS in 7m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f802d37e8db04a64a10bc00663c6a85a : SUCCESS in 10m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/093af382538247afb68b7d36426dd666 : SUCCESS in 13m 20s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/021942c7d5bf4b859ccd4257f61c1b10 : SUCCESS in 6m 08s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/b628294673934330ae18cf11247abe61 : SUCCESS in 28m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/0c0bb846a0604a2f8eeb6c6b20c39014 : SUCCESS in 12m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b688bd3c42584020b75d3315995f4263 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e7ae4cdc87ae4f208fc8f79c3455e6e1 : SUCCESS in 16m 24s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/0066e1c6a34f4b00bb5f560844d5bfe6 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4f496b68974444fdbe4633b13bd39c31 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/efb6f202e11e496f87ddd753f64eae40 : SUCCESS in 16m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ca959dfe12b40e3b0ee752c74c53ac5 : SUCCESS in 53m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b9c775f70674f6988bd5a1f4f494d97 : SUCCESS in 1h 46m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2f93249c299b443618dfd47f9a02f83d92274909","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-05 23:34:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5d0662384ec8063c334c31069db3590a68dca9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 23:35:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0aa469dbefcaee4dee693272d707bc286f8729a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 04:42:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7591cb053284ef19536c73b4c3f4d23 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ccc64b9c8dc844338aaa08ae4d6d3815 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b818d986b136429298a3672275621109 : SUCCESS in 5m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0d7d10c229d40d3abc8d602c24d9ed8 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c47365f55e7418997283e06d25637ed : SUCCESS in 4m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/255936dac6d0495b9315f38d0f60099c : SUCCESS in 9m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e5eabd90d17b4014ba5153b80753a92a : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9a0d929907b7495bb0d51ced9396f923 : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/dfe3d0d453104249b5f74c39260a2364 : SUCCESS in 6m 03s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/245875a8d88746369d20068b11523a4a : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06b7a1e8a84d4bfb81986f71c4fc23ce : SUCCESS in 8m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9a5917613b5048cbac427cb70935b8a8 : SUCCESS in 11m 44s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/83fb4cd338c04fcf949f44a19f0dccb0 : SUCCESS in 14m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/fcf6670b46d24ca2bb4240c0b50bf4b2 : SUCCESS in 5m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f6f26d7ea1d04c76bde0a6b77f3f7902 : SUCCESS in 23m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/78fc8d1431404aa7b07ebc7a4d80e1bb : SUCCESS in 12m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/48c2aaf8e97f46c583fbb00a0ad753cb : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a227e14572854476aad7e16fde9708fc : SUCCESS in 15m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ed8d338a1edc40ceaf17280bcb7cdd0f : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e3e746e6d4443b39a0e543622e5d434 : SUCCESS in 7m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/c5f4aa72ab5e4dea8f1c28aa90efc793 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/627fcdaa3dbe4274b96e4e8d0613f9c9 : SUCCESS in 50m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a03f31e37628484984d3270b622a7f53 : SUCCESS in 1h 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dc29679beca1d903ffd3685ef09bf1e89fecb7f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 04:42:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01252901efb895ed370a86f4ebd6547d770dd285","revisions":{"01252901efb895ed370a86f4ebd6547d770dd285":{"kind":"REWORK","_number":1,"created":"2021-05-05 15:50:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/789907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/789907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/789907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/789907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/789907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/789907/1"}}},"commit":{"parents":[{"commit":"a34c9d00c90dbfe5dfc30d8b61db69ee16c532da","subject":"Remove Xenial wheel build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a34c9d00c90dbfe5dfc30d8b61db69ee16c532da"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-05 15:50:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-05 15:50:52.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 15.1.0","message":"update constraint for tripleo-common to new release 15.1.0\n\nmeta: version: 15.1.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Ia064530696cb5c47b009967f665900c2dd06d44e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I1d4cf4e9d00fa35e7a324f927c929e8720bd13c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01252901efb895ed370a86f4ebd6547d770dd285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/01252901efb895ed370a86f4ebd6547d770dd285"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
