)]}'
{"id":"openstack%2Frequirements~790091","triplet_id":"openstack%2Frequirements~master~I1ef4aead6a554448ec9c557961f8391d6493ab87","project":"openstack/requirements","branch":"master","hashtags":[],"change_id":"I1ef4aead6a554448ec9c557961f8391d6493ab87","subject":"Remove tripleo-common from global-requirements","status":"MERGED","created":"2021-05-06 15:52:47.000000000","updated":"2021-05-12 15:55:56.000000000","submitted":"2021-05-10 15:11:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790091","meta_rev_id":"253034c6d947f70cdcf1fb9b61c7531bc6e021ea","_number":790091,"virtual_id_number":790091,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-10 15:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-09 04:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-10 05:46:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-10 13:56:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 17:10:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-09 04:23:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-05-10 05:46:51.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-05-12 10:01:53.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2021-05-12 15:22:40.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"}],"messages":[{"id":"15aefd1da6dbdea99f671deb12a5251ccfae9af5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-06 15:52:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c279aa2e8b7aa665449337f1a5139cb9fb50561f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 17:10:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6047ec585b0b4cf3938d3c51fc75949a : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecec066f1b2b41d0b13e750f60f21e45 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af2f3408129040dcbf733f1bcba2e056 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5058c019b1d84e35891954a08aaea42c : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ca90b4303634e7d8fd02c8a62fcd587 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7c93f63c6374cc79c526776510e36b9 : SUCCESS in 4m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1dc66f3f3fa429488a9f6d6ffa56c99 : SUCCESS in 44m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50449ddce311466c9c904ae6a64d4964 : SUCCESS in 1h 15m 49s","accounts_in_message":[],"_revision_number":1},{"id":"0836212decf349b1506db622c9bded800028f7e1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-09 04:23:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9650c13e92e86ba0b5e8e21a54d64aff454037d9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-10 05:46:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"76d9319e555414e75932c8b76c26872b5432ad0b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-10 13:56:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ec4c1105fcb0e7b973f743a6fd50aea76214212b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 13:56:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0e084aee4853528fc4884b4bfc91390ae9a35745","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 15:11:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a760bba18164be695ba803d7935d610 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dba8f1bb515461188f6cf0fd7744cd7 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2f7f1b62b2a4a50a89d496208d921fb : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0473548e816a46649feb7435f3dbf624 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f070501c67f14ac38ebed5b5be0aed90 : SUCCESS in 4m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb91ccefc09a4f95ad3870387e27fce4 : SUCCESS in 46m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/487f4af81d9c4f70aeb7ec0acbd046ca : SUCCESS in 1h 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a7341defa40f4487f32e229f181c13be272f7b5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 15:11:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"caeb3f0d36b777bdf36b751a01903b6352fa695b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 15:13:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4079e5b58a548d5b2e72b8a60ff4177 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ea019da4216860218eec6a6247adf704baafa932","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-12 10:01:53.000000000","message":"Patch Set 1:\n\nThis has broken all tripleo requirements-check jobs. FWIW https://bugs.launchpad.net/tripleo/+bug/1928190","accounts_in_message":[],"_revision_number":1},{"id":"e88f895813f7f8073053886617aa8c8a3b2ae8f2","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-12 15:22:40.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e This has broken all tripleo requirements-check jobs. FWIW https://bugs.launchpad.net/tripleo/+bug/1928190\n\njust got some time to dig a bit more there... added a comment at https://bugs.launchpad.net/tripleo/+bug/1928190/comments/4\n\nregardless of the eventual outcome, I think we need to revert this now to unblock the tripleoclient gate","accounts_in_message":[],"_revision_number":1},{"id":"253034c6d947f70cdcf1fb9b61c7531bc6e021ea","tag":"autogenerated:gerrit:revert","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-05-12 15:55:56.000000000","message":"Created a revert of this change as I4dc6f68fabdab8be21e47135473bd61134aa0717","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7e30166d07bd7d3d9bfac72ed64ccfdc7edc4cb7","revisions":{"7e30166d07bd7d3d9bfac72ed64ccfdc7edc4cb7":{"kind":"REWORK","_number":1,"created":"2021-05-06 15:52:47.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/91/790091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/91/790091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/790091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/790091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/91/790091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/91/790091/1"}}},"commit":{"parents":[{"commit":"7277d40e84558df20263dae48a6f52f4e718b67c","subject":"Remove Xenial wheel build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7277d40e84558df20263dae48a6f52f4e718b67c"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-06 15:37:57.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-05-06 15:52:39.000000000","tz":-240},"subject":"Remove tripleo-common from global-requirements","message":"Remove tripleo-common from global-requirements\n\nAs of Wallaby, TripleO no longer requires Mistral for deployment. The\nonly other projects relying on tripleo-common are within the TripleO\nproject itself and those projects are not in global-requirements. As\nsuch, tripleo-common can be removed as there are no dependencies with\nany other non-TripleO OpenStack deliverables.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: I1ef4aead6a554448ec9c557961f8391d6493ab87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e30166d07bd7d3d9bfac72ed64ccfdc7edc4cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e30166d07bd7d3d9bfac72ed64ccfdc7edc4cb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
