)]}'
{"id":"openstack%2Frequirements~790389","triplet_id":"openstack%2Frequirements~stable%2Fussuri~I323d0e2d5a924f58c2c18531ca61e578f43a64d0","project":"openstack/requirements","branch":"stable/ussuri","topic":"new-release","hashtags":[],"change_id":"I323d0e2d5a924f58c2c18531ca61e578f43a64d0","subject":"update constraint for neutron to new release 16.3.2","status":"MERGED","created":"2021-05-10 09:33:21.000000000","updated":"2021-05-11 18:39:47.000000000","submitted":"2021-05-11 18:39:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"790389-new-release","meta_rev_id":"de16f1d9b081f622ccfe87709739c0bac5d2a639","_number":790389,"virtual_id_number":790389,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-11 18:39:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-11 17:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-11 17:10:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-10 09:35:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-10 11:03:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 13:56:22.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-05-11 15:44:40.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4d639e72990fddf42b9977f5e5e60bc041209cca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-10 09:33:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"782c795273c0e0f5b6d11c0137d5f08c553a80c7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-10 09:35:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afac4ee44cfb5e82a28eb0c415e1c215b705c867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 11:03:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07117636f58f4b9bac54bfc4a2dd9619 : FAILURE in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a6cfe0b23f941ccb9f1fc2efe63f887 : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/31154ba5b6584bdf814d7ce4d5510814 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8fc6f5434974f1db4b99854e3b880ba : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3431f0b765ba4e7ea30c4a56cfbb7095 : SUCCESS in 3m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/484d6fbfbfc9441ba375d2f59554a4fc : SUCCESS in 9m 56s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/309b47f384894fd6b99931cfc310ce96 : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/572a3a1ea7164d93aa74e352e118b4dd : SUCCESS in 5m 13s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2aee7ba39884451ebc7e18722c35a5ae : SUCCESS in 11m 42s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a33402b18784320b03d27f13691549a : SUCCESS in 7m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d5a2444e1df4457c95669853bd260a3f : SUCCESS in 13m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/6829e6386b3e4115b748829f59672a78 : SUCCESS in 27m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/903290447c854940b1e85d40c913e9dc : SUCCESS in 12m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/d122048456a84f4ba03ad6829791fb60 : SUCCESS in 13m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9d0190e6c52049ae82ef75e606c20f13 : SUCCESS in 19m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f9dd840a04504f22b52b1039409721d0 : SUCCESS in 5m 23s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/e20a3d6cd6ca4ef7803aca1a40bd56ee : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/710e0cd94d6943e4bf5ff6603e79a51b : SUCCESS in 6m 35s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/976b355c391d4f8dbc0e6721b44eea2b : SUCCESS in 10m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a18d9b35452241daa54ee07f14e4a51e : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd2933fe8baa4a84a1040894a05d292d : SUCCESS in 58m 51s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ac6b38b150641809e2be54639212339 : FAILURE in 57m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d87f97c5ce24e0588cd76fda95ce591 : SUCCESS in 1h 24m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/83904daf150848f08bd7b4a016855b14 : SUCCESS in 1h 07m 04s\n\nWarning:\n  Comments left for invalid file openstack_requirements/tests/test_check_constraints.py\n  Comments left for invalid file openstack_requirements/tests/test_update.py","accounts_in_message":[],"_revision_number":1},{"id":"31bb3e70d1534851f07e581c9e025cef7f9aa377","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-10 13:56:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b2d25763e6a2c5269aff1f35b89fb1b36a7d2d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 15:48:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b81a5cb45247c1940b2af9b69875ae : FAILURE in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8308cd0af0d644ce85c8d9a34996c9c2 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4caf7a839b94d519c3cde80db5f6a3a : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/206a47bb68324c0dad97257436c46ec8 : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bd222b30d5fe4da7b6e8ac94727f278a : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/50bbc5489b2743ce9a479687ac3bfd21 : SUCCESS in 10m 10s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/59f7105d47014a2293b10e220a16ffaa : SUCCESS in 13m 33s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/68d84633e51b4b4facfca3f80d117cf0 : SUCCESS in 6m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/bc0521bf835944289dc8efed334210d6 : SUCCESS in 11m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/68a82129413b4a54a4b2bb18c456f29c : SUCCESS in 7m 29s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/38a0a39fb54e4a6a8ef12d27c0587ac8 : SUCCESS in 25m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/407705873fff410695fb54e4c4265b8b : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4f351e55e4c54b11ac4e7c91f2b39c29 : SUCCESS in 13m 33s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/451a007d28c04e13bd88186ce877a384 : SUCCESS in 14m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b0945ec465814154b9d004013fff93ce : SUCCESS in 18m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce8fb11f76ff410f8b6e39923261ca5c : SUCCESS in 7m 28s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/dca3912f3b9f4bac98e9a181eb785bb2 : SUCCESS in 7m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/40ceafa353214214a5860fe6ea05ab58 : SUCCESS in 6m 22s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/c0cdbe5c949849ad84e1bfbe63e287a1 : SUCCESS in 10m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/48113b6cd55944a8a203d87ff13f0373 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91dd81bcb95343c789b632078dd836ff : SUCCESS in 50m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9499d6419d6d47348dc54a5df329938e : SUCCESS in 1h 22m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8621c148c6d4c10a8821c168e762005 : SUCCESS in 1h 45m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/7ca8aa211ae54ed88ca6462eb32ff05b : SUCCESS in 1h 04m 14s\n\nWarning:\n  Comments left for invalid file openstack_requirements/tests/test_check_constraints.py\n  Comments left for invalid file openstack_requirements/tests/test_update.py","accounts_in_message":[],"_revision_number":1},{"id":"82b499f822c7d0ba8516caf8f4abdbc4befeacbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-11 15:44:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ef4a8c51d7f00d116a166b602d45f3d7491825da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:00:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908a838cfe2a46379c6444677aeadf5f : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0386b5d229404fc7a4d0eefa3ec1cfab : SUCCESS in 6m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f8ab1d575dc4ff389374ff82ef98cf6 : SUCCESS in 5m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e65d3153782948c5b7bc7ef9f411833a : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/928c6359a39e4d8ead8fb3cc19ad8bc2 : SUCCESS in 5m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5dcd4b88fb1d41699d0446b417237e32 : SUCCESS in 8m 58s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ffbb487ac62241f4b9c0e2c6bd510613 : SUCCESS in 10m 31s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/212c53dd82d74586a2f73bd79fd73e0f : SUCCESS in 5m 18s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c88f4636ad514156800947958f89ff40 : SUCCESS in 12m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7920de531727405a93b2431b1dab3cef : SUCCESS in 10m 48s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/342bb562c43547b4ae3bc97ecb976cd6 : SUCCESS in 17m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c4aa7ae1cb974cc89f8cfef7805b48ac : SUCCESS in 31m 37s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/9727495738e94fb78426f536bae0d968 : SUCCESS in 14m 27s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f208b86f33464a3ca49316627f353900 : SUCCESS in 15m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0111b690658d4fe9a860605333c431d6 : SUCCESS in 16m 55s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/2c0807ef05c94327b96748f456c5fb9f : SUCCESS in 6m 02s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/ad0e7fe630cf42cd9e6e3a4c577ecc7b : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/269a07bee73e4d6a8431f320b6d097bf : SUCCESS in 7m 16s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/b8ea20d559c74a96a1236495c4e01cb0 : SUCCESS in 12m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/22fe0ea40cc74d9c9113681dd697aa13 : SUCCESS in 12m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9d7c1c24e604dd8a07f655d75c415f9 : SUCCESS in 56m 40s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/28e369ee291c491890e8937319dbf86f : SUCCESS in 1h 11m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85e398b8a4ba4631b3bfce070acf1082 : SUCCESS in 1h 12m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/dd37220abc32476299413222b9a78985 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":2},{"id":"98f2dad94ca47a966c926ae304ae463a25482a4e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-11 17:10:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"03a984e3a4aaa843c514a8dded30dd29b982dbb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 17:11:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f62c0957feb3ce1fbba064458aa8c55f2e0d8b73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:39:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/705e26d474e948ed9724e59a95b6aab0 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c47f940993a47daa316fbd5e0a1db04 : SUCCESS in 4m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e2c8f61b6589420c85fa8a3d90876d52 : SUCCESS in 4m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8f2ca0a031b48749122f954861d3bc9 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/57f94a9a334d47e99d71cebf23ef82c8 : SUCCESS in 4m 04s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7c1633dbfb334b0faf7f97fff71e3538 : SUCCESS in 9m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/04d811ff5eae40f198c87e09e276809f : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/57d1d16f31064814a0dfa4c9579ed02a : SUCCESS in 5m 25s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5be4a0d328354b95bb7bd018df0cb5e7 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/879b07b4663745e9804f04dde59144ae : SUCCESS in 10m 40s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1e947eb1eea147b6a41a85837451759d : SUCCESS in 21m 23s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ea9d50f1fe9547dca1169a16fabc8a14 : SUCCESS in 33m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/22efce6263124e3bb2af8f3d4f315af1 : SUCCESS in 13m 09s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/f91020a0d17740f294861a4e9ee57a35 : SUCCESS in 29m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/903469effa6b4ef09d9e2031ee574ed7 : SUCCESS in 23m 26s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/978e5150dde04af0839520b4bdd44afd : SUCCESS in 5m 31s\n- cross-octavia-py37 https://zuul.opendev.org/t/openstack/build/104944fac32043ed9445e2a6cf0da069 : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6d9b44a7c7e241ea81002b86aa8f0c1c : SUCCESS in 9m 34s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/86c978070c5643dfbf0db5425fef5c0f : SUCCESS in 11m 05s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5d7117247e4c4d53b45feae6f78a2771 : SUCCESS in 10m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/135665a166a049ed8734765aa3d9a1e5 : SUCCESS in 56m 13s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/c033ccc2acda4f62a62fc1153c1af20b : SUCCESS in 59m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7bba60e54fea41099a669fa72162ee0b : SUCCESS in 58m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/b26a491930634401ac4bedb3811012f9 : SUCCESS in 1h 11m 27s","accounts_in_message":[],"_revision_number":2},{"id":"de16f1d9b081f622ccfe87709739c0bac5d2a639","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 18:39:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"485a764df887e74e348674da1aeb461ab6feaaf3","revisions":{"16ea18761f5d71e483d91a4e9491f7b548313612":{"kind":"REWORK","_number":1,"created":"2021-05-10 09:33:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/89/790389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/790389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/790389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/790389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/790389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/790389/1"}}},"commit":{"parents":[{"commit":"a69a8d1550f38216c73e8a0924b2b65bed785685","subject":"update constraint for sushy to new release 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a69a8d1550f38216c73e8a0924b2b65bed785685"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-10 09:33:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-10 09:33:15.000000000","tz":0},"subject":"update constraint for neutron to new release 16.3.2","message":"update constraint for neutron to new release 16.3.2\n\nmeta: version: 16.3.2\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Commit: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Change-Id: I1f0303eceb9de95de443f999a43d048f31e5f0ce\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: likui \u003clikui@yovole.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I323d0e2d5a924f58c2c18531ca61e578f43a64d0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16ea18761f5d71e483d91a4e9491f7b548313612"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/16ea18761f5d71e483d91a4e9491f7b548313612"}]},"branch":"refs/heads/stable/ussuri"},"485a764df887e74e348674da1aeb461ab6feaaf3":{"kind":"REWORK","_number":2,"created":"2021-05-11 15:44:40.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/89/790389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/89/790389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/790389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/790389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/89/790389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/89/790389/2"}}},"commit":{"parents":[{"commit":"a69a8d1550f38216c73e8a0924b2b65bed785685","subject":"update constraint for sushy to new release 3.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a69a8d1550f38216c73e8a0924b2b65bed785685"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-10 09:33:15.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-11 15:44:09.000000000","tz":-300},"subject":"update constraint for neutron to new release 16.3.2","message":"update constraint for neutron to new release 16.3.2\n\nalso switch to unittest.mock to make flake8 happy\n\nmeta: version: 16.3.2\nmeta: diff-start: -\nmeta: series: ussuri\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Commit: Bernard Cafarelli \u003cbcafarel@redhat.com\u003e\nmeta: release:Change-Id: I1f0303eceb9de95de443f999a43d048f31e5f0ce\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: likui \u003clikui@yovole.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I323d0e2d5a924f58c2c18531ca61e578f43a64d0\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/485a764df887e74e348674da1aeb461ab6feaaf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/485a764df887e74e348674da1aeb461ab6feaaf3"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
