)]}'
{"id":"openstack%2Frequirements~791020","triplet_id":"openstack%2Frequirements~stable%2Ftrain~Ifb57dd73128ece6fe6898496bb38ae834c0a3d5b","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"Ifb57dd73128ece6fe6898496bb38ae834c0a3d5b","subject":"update constraint for os-brick to new release 2.10.7","status":"MERGED","created":"2021-05-12 16:13:59.000000000","updated":"2021-05-12 22:26:34.000000000","submitted":"2021-05-12 22:26:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791020-new-release","meta_rev_id":"65a6f38390ec7f25a2c777e1523b5330393f783b","_number":791020,"virtual_id_number":791020,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 22:26:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-12 20:35:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-12 20:35:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 17:42:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 19:02:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-05-12 20:35:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ee324ea580a6596b0f26cab0dcebc029bbef2c15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-12 16:13:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d921f1743803d631d8a7028d0f0abc8ed7102a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 17:42:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61175eff9c4b4b6aa7b3d564a0a16654 : SUCCESS in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1d2f52961319472daa5dfee46ef34e5f : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cc69f3f20334615967e57bffdc7982e : SUCCESS in 16m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94fe2bd846a84a319dfd109dde8e51d1 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11adf15e5b5e40898e0498619f8d6ab5 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d7360f7ebe7f4d2cbf255e7f50104572 : SUCCESS in 3m 31s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/bfd51f6d791e447eac9d13c71e8ba699 : SUCCESS in 7m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a8fadb13b551449cacda8c8b6841d0fd : SUCCESS in 9m 41s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/377bb72971fe4a8e98d9408df8b07ce0 : SUCCESS in 11m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/3c5371c88032495fa248098a719d27e4 : SUCCESS in 9m 18s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/cbf4583a3961434992f612ee8304c5b1 : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/120de560191e4fa5bbd99dc58baa21e5 : SUCCESS in 5m 08s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/67ed2b905eba42d3944908a44ccb39ba : SUCCESS in 11m 10s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/46b836d2cc304c5cac793b7f4908bb47 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d44d05e6d1e943f7bcc3284a4a08fdf2 : SUCCESS in 7m 26s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/af13dafe86114ea59198d3e8e0137e78 : SUCCESS in 15m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bde5714ece554e0ca4ebce8ae468fb71 : SUCCESS in 17m 00s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/a8d83cad229f4d7eb5efa8fec0fea4db : SUCCESS in 21m 47s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3b39c39c02d6480e8f50f73f5fcf1be5 : SUCCESS in 23m 19s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5585e0cbc6314c8d9f2857487cbfefb3 : SUCCESS in 12m 27s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4d3334df046347e2bba345e053a7a0f2 : SUCCESS in 12m 36s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/a3680b754969402c94cbcd65929debb2 : SUCCESS in 12m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dfe79a136ac34cb5bb5fe554fffee2c1 : FAILURE in 4m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8151915ddca94dd1b8685196e3c086e9 : SUCCESS in 6m 03s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/2a82410aba5b4661a4478e98ff327641 : SUCCESS in 12m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8787661e1888461494e024a5f014f4a6 : SUCCESS in 11m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b56998bd75934aba9091806b8ad533bd : SUCCESS in 57m 33s\n- tempest-full https://zuul.opendev.org/t/openstack/build/e03f789a78b7491f964821bff38280c1 : SUCCESS in 1h 26m 47s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/2927ba4fd04247e79e5b349ced889a15 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"3dfec0398e82a94c8afab3a80fa9ffcd7de9acb5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-12 19:02:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e0de7bfc47df9fab17c9e8c439bd9eae265f0ebf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 20:33:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/602b21be82654ab5b8bc6115db621abd : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/388d5eaf17ee47b8a311006af66f8988 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ef39606a304c428256c03fd92246d7 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b5006e4ccabe4352b9f6c4f9a5cfb9c5 : SUCCESS in 6m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/af857bf1e99e40d08f2ca9ed5c3361f2 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fcd1738a8402482d8e19e7d9871b364d : SUCCESS in 4m 39s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/290297ed15e44066b3e474a20cfa896d : SUCCESS in 8m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/11f4ce2ffd8b46619bb26b6394da06f1 : SUCCESS in 10m 21s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/c46cc9df1fd34032a357b5744fee49cd : SUCCESS in 10m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/c5b133504edc4b638376a33946a68291 : SUCCESS in 9m 55s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/1b03aa0b44ed4de09184d7575e36fd0d : SUCCESS in 5m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/9cd16eedfb99496eafa619df76602166 : SUCCESS in 6m 10s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/803b0b9a957349ce9dbe3588dfb33482 : SUCCESS in 10m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/7211e2b7e116464c85b3764b09030ee0 : SUCCESS in 9m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da4489934d9b4d91b085e0b781ea81ce : SUCCESS in 8m 35s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/79d7d7504368412b8ce22d338a11a648 : SUCCESS in 19m 26s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/76f43d253fec4a1f9020b719cca29ff2 : SUCCESS in 18m 36s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/aedcee2c045548e3a6e4402c755b0fa8 : SUCCESS in 23m 17s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ae3a2ab798db40829d48c2740bee2efd : SUCCESS in 15m 37s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d10f86100d384699b2ee2e35cd031fb8 : SUCCESS in 12m 18s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/f742220b385746299fc4519095e2cdce : SUCCESS in 12m 22s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/fdf304837e4043c3b2b5240fb947b8e4 : SUCCESS in 14m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8559b0964c204e4eba1386dad750dca1 : SUCCESS in 19m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ed39b0ca63c643b5bdaf114d29209651 : SUCCESS in 6m 26s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/24bed12d895b4efa83cd79fdd9466e45 : SUCCESS in 11m 46s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bfd543264ee54748b75bc6d7971ccba5 : SUCCESS in 11m 35s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fd875a419e57465eb9933c548b2972cf : SUCCESS in 54m 01s\n- tempest-full https://zuul.opendev.org/t/openstack/build/636982bdf8a740d490ffd57cfb3bb27b : SUCCESS in 1h 24m 36s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/fc8129b64113483f8070c4d47eaea926 : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2e5cb8fff3c9509fbd8905034695d73b0bfbd9fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-12 20:35:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4692457ef0393bff61a4dbd82116a110848feaf1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 20:36:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e8d156a4da904d09f3da1b4f05985490c6bc5255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 22:26:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/102808fe43264a45b0990e53902a31eb : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd5e1b553f38408a9b92dd63dc929a37 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/681fdfa3c972449dae4b7a9790805b15 : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/679dfc1e220d452ead2ad2ef6cfb1cec : SUCCESS in 4m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7b3d6aaf416c4b71a084ef239377f802 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cb35043450424c6386e4655130311b54 : SUCCESS in 3m 47s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/0d52c824b32442c6ab347c0cea54431b : SUCCESS in 7m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/863244ee2b2e4222b0a292aa49356106 : SUCCESS in 9m 32s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/2c5d1410d5da4d609aa9331739a1116e : SUCCESS in 9m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8b07ad919cd44c4ba2f2c6e2d6982671 : SUCCESS in 10m 32s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/ddd5571b24a54c59a4fb1b4552bbb240 : SUCCESS in 5m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c666a759633b4887b23a3012e7d3562e : SUCCESS in 5m 07s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/12135265eef4456f8c06effef4e0b839 : SUCCESS in 11m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d39f0b69fa2a4fd2aedc9ca6decb729d : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fdfcc7d728884200a0b121aa829b99e6 : SUCCESS in 7m 42s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/c8f235f0fda74036a1281cdecfacccdb : SUCCESS in 15m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fec4b0740e524adbae6a41f8cc9148be : SUCCESS in 13m 25s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/fc85d06e0a09444fa7d86f4b4abbf1f9 : SUCCESS in 23m 43s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fb143bb31d344a0499c357682061f0ae : SUCCESS in 20m 25s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/5e94489993a24461b1b0ec4c2d3b1798 : SUCCESS in 14m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/df0b40f7c6954b4fae0ba87150cb3fa9 : SUCCESS in 14m 02s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c118a04ab58d4e9e857a8d2448df21ce : SUCCESS in 14m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c7ac171bff5649249819211937b185ad : SUCCESS in 20m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6814fc47520e42f385b32b7d4de55002 : SUCCESS in 6m 07s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/3a60c6fbd4a64e29bbfaa1a584a7300c : SUCCESS in 9m 17s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/73a6195eea63470ca372fa3380449884 : SUCCESS in 11m 50s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/098f6d9ea3204f69a350112f52c10cd1 : SUCCESS in 52m 53s\n- tempest-full https://zuul.opendev.org/t/openstack/build/0364833a421f40c3b68f8c725810ba51 : SUCCESS in 1h 42m 13s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/bf9a704de0eb44f9b8917e61aa786ebc : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":1},{"id":"65a6f38390ec7f25a2c777e1523b5330393f783b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 22:26:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b169132f0268e12c28486a3cf5eaece0c974024d","revisions":{"b169132f0268e12c28486a3cf5eaece0c974024d":{"kind":"REWORK","_number":1,"created":"2021-05-12 16:13:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/20/791020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/791020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/791020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/791020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/791020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/791020/1"}}},"commit":{"parents":[{"commit":"335a3ffbf5222cea2b4e31ae60bf99a04e2009c1","subject":"update constraint for neutron to new release 15.3.4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/335a3ffbf5222cea2b4e31ae60bf99a04e2009c1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:13:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:13:56.000000000","tz":0},"subject":"update constraint for os-brick to new release 2.10.7","message":"update constraint for os-brick to new release 2.10.7\n\nmeta: version: 2.10.7\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Commit: Brian Rosmaita \u003crosmaita.fossdev@gmail.com\u003e\nmeta: release:Change-Id: I50e23a0f21af8d73de09029da543fb6b975f60c9\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+1: Jay Bryant \u003cjungleboyj@electronicjungle.net\u003e\nmeta: release:Workflow+1: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nChange-Id: Ifb57dd73128ece6fe6898496bb38ae834c0a3d5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b169132f0268e12c28486a3cf5eaece0c974024d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b169132f0268e12c28486a3cf5eaece0c974024d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
