)]}'
{"id":"openstack%2Frequirements~791028","triplet_id":"openstack%2Frequirements~master~I61831c6d92904fa3b0ed81c66681a9b9dd651fd1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I61831c6d92904fa3b0ed81c66681a9b9dd651fd1","subject":"update constraint for oslo.vmware to new release 3.9.0","status":"MERGED","created":"2021-05-12 16:53:06.000000000","updated":"2021-05-12 21:31:11.000000000","submitted":"2021-05-12 21:29:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791028-new-release","meta_rev_id":"e6a05c810073ce0c20391b413daff63879db5d11","_number":791028,"virtual_id_number":791028,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 21:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-12 19:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-12 19:02:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 17:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 18:04:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 19:02:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"439cf9fd073f964e3b5c1754e7b935309369eba2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-12 16:53:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33aa4b6b62c01b0416f838dd3da37068586f91fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 17:09:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cdb5221143ab47f896037269a9f11a11 : FAILURE in 14m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e44d35493be67dd92239eb5f899e3d889ed905ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 18:04:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b98d6e397aa4a3899bfda28253812c6 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc5ab02f68874895aea640b77fa6ed24 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e44cbd3a56654378b98afa8d1e7928f1 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5ac995561c9459daed104ef2ddf15f1 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43223291633044ed925c984c970d8665 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3405aa7c0b954386a83ea0997bea7064 : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92de008eb575493fa4640c0147f93b4c : SUCCESS in 3m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bbc90d77464b4100a04bd77ca495ebb7 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/771b147ce8024d52a0ad9a7bc6404bcb : SUCCESS in 8m 39s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/81c719bc6ce34ac4840cec2d1c2676d7 : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ae54c9859a3649d0a6ac06200e6fcab7 : SUCCESS in 10m 05s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/9ef42c6e77fb485bbd79389d7f22beb0 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7ca7afab1ad4c92b738a6bad45b4d97 : SUCCESS in 7m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ad03b4838ea940208aff91ad97e506bd : SUCCESS in 21m 08s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/2ce9e37ff00d455e916a8622553f2cf3 : SUCCESS in 15m 23s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/71eb912a3dcd4dceb231f94a297e093f : SUCCESS in 5m 12s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d16cafcaf2584121a6b4d51e0731bcf1 : SUCCESS in 25m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/993ff1037ab44d97857cbbeb2b06e1d4 : SUCCESS in 12m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/56aaa33ee3eb45afb750867c09df5526 : SUCCESS in 4m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/321d90e02be94442b902e430a06e2c35 : SUCCESS in 18m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f2f5a47725414f87925fba815ce53387 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdfe0c83e1324875af393cf184a03425 : SUCCESS in 6m 36s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a0f643c1e4e14b898f842b4399260648 : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f6bd4cb29d347f580e8ca865d414717 : SUCCESS in 40m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f689c679eefd4066bae3b0861d3b4f4e : SUCCESS in 1h 10m 41s","accounts_in_message":[],"_revision_number":1},{"id":"10b1b38d6b3f461190f7e8e3eb4e1f3436905d1c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-12 19:02:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c8c6c7394c6a2ef0223da9002a074847cbb55e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 19:03:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a1d053c1fd62393b9635c43b96e646e85b7d027","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 21:29:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced87b5657374432908f58ff490d0f90 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d111514d36fc4dedb0ec26aed8960535 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2464d2409e740dea9a0d51582a32c2e : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9763c9492f24cf48f87cd8175d2c268 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b08bc03d10bc4857b038d8638ec9060c : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d419efd053224e6d976d2e13e8d2872e : SUCCESS in 5m 58s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a54531904f94b78aa87e879255f57bd : SUCCESS in 9m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cbc8f2c02ca94a9da1131b5756de5578 : SUCCESS in 9m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d53772decf8943a1aad0366417aff7dd : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf75f19765cc4892bdc4d9d093ba7bb3 : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/a126904b00f4477aa278e4f3b379a8ba : SUCCESS in 11m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1ea8345cb332402cad965fd33edf3e68 : SUCCESS in 7m 50s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/04b58b345b9b48a7ba0928998375a075 : SUCCESS in 16m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a972107c06db4de881d5234160274e2c : SUCCESS in 13m 35s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/4bd8726c9b994a2b9d06f0a2bed36762 : SUCCESS in 5m 26s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fa44dd5834964896b5e53533ccfd74db : SUCCESS in 30m 10s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b00f114a96f8481b85d2aba4bfb72d8b : SUCCESS in 13m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5815ff991e284cfcb5e95492a5dab142 : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/93ec3b05555a433aab289010e1eeab54 : SUCCESS in 14m 28s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/065af4f0b8a6498f890b94cee2623ba6 : SUCCESS in 5m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7e7ac8ab9b144569022b8c1dc963290 : SUCCESS in 7m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/6ba00f60056f46f290995da50288b4c3 : SUCCESS in 14m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc874cf2308742e5bb08054407a6b266 : SUCCESS in 45m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0d2de135a38477a869c5959348f1dbc : SUCCESS in 1h 39m 07s","accounts_in_message":[],"_revision_number":1},{"id":"79773e683aa5e07ea41ba1497e2091e6dd478ea4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 21:29:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e6a05c810073ce0c20391b413daff63879db5d11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 21:31:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8df3dd9ad27242ffb245e28eeb9b9ccf : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f35669b401cb15398dce8097af6821990ed9f653","revisions":{"f35669b401cb15398dce8097af6821990ed9f653":{"kind":"REWORK","_number":1,"created":"2021-05-12 16:53:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/791028/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/791028/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/791028/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/791028/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/791028/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/791028/1"}}},"commit":{"parents":[{"commit":"ce19462764940a4ce99dae4ac2ec7a004c68e9a4","subject":"Merge \"Add python-linstor to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce19462764940a4ce99dae4ac2ec7a004c68e9a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:53:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:53:03.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 3.9.0","message":"update constraint for oslo.vmware to new release 3.9.0\n\nmeta: version: 3.9.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iaaeeb0b029db04958eaa3c5399ae1580a8ea2914\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I61831c6d92904fa3b0ed81c66681a9b9dd651fd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f35669b401cb15398dce8097af6821990ed9f653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f35669b401cb15398dce8097af6821990ed9f653"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
