)]}'
{"id":"openstack%2Frequirements~791029","triplet_id":"openstack%2Frequirements~master~Ic72017165fbc7fb47b2e7ca53436bc91539c2fde","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic72017165fbc7fb47b2e7ca53436bc91539c2fde","subject":"update constraint for oslo.db to new release 9.0.0","status":"MERGED","created":"2021-05-12 16:55:10.000000000","updated":"2021-05-12 22:03:24.000000000","submitted":"2021-05-12 22:01:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791029-new-release","meta_rev_id":"239eff5ef04dabf9b08b77a362504fe2395aac75","_number":791029,"virtual_id_number":791029,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 22:01:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-12 20:32:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-12 20:32:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 17:11:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 18:16:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 19:02:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-05-12 20:32:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f98413bd3e2aa989ff2a5a557935dd5d6ee3241a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-12 16:55:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73dcd5ebdb9993f1ff980ed42790500a3de057ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 17:11:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1da4ce66703c4865bf5fb763fb2c32b8 : FAILURE in 10m 48s","accounts_in_message":[],"_revision_number":1},{"id":"c1324cd44822b7618fd0e7d2247aa3c58aed5211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 18:16:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0717cfded242abb1f57b860a59ec60 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e37e1e3dcd546e8a3df06cf055d6609 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d927b7f8de55445ea6a26f3221d7a705 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2edafbce698340849bfe166e4663635b : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d433375324f3461d8d7757f24f29e95a : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42ac72aed8c5431dbb0999a454c23fe7 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/54eb4ddb088c4422bc3ae991c754f6b7 : SUCCESS in 3m 49s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1298c5581fe24614836dc0ea10d903d7 : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4dd7d5406ef04f39bc9fbc2eedf6bfa4 : SUCCESS in 9m 06s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9604f87b665b4b17a4fa034878d86d30 : SUCCESS in 8m 07s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6338cdc91b0648529a89cf5de59fedf3 : SUCCESS in 5m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/60720043f5fa4106992eb8f2d081d0ca : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1284ffee68de4c9ab1448abc3fdbc73d : SUCCESS in 7m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bc09e5772c1c4922a37490fb029cfb0e : SUCCESS in 13m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/8f6b9b14137a4af380a2b02e6833e4f4 : SUCCESS in 11m 17s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3fd89e39d6914b2496a4b71de588e70a : SUCCESS in 5m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/248624ba78fd4ee09d70d7c6d61155a4 : TIMED_OUT in 41m 04s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/20e7e8a91dc2424ea96fafcbf74226a1 : SUCCESS in 12m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab261f68a21e44b0ba74d9359032b126 : SUCCESS in 5m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bbb85a905cdb4998bdf7d9798e414bda : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/96abd3281ef74fcdbc03fa8ef7d847ed : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b0a28985763146a09b09ea1e2b9357f0 : SUCCESS in 7m 42s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/51564f0edc964c049db0f7e37ce15430 : SUCCESS in 12m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bb5040df41d404cb1b2419227c72469 : SUCCESS in 50m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f03dc74310564638a5e2162481672f2a : SUCCESS in 1h 20m 19s","accounts_in_message":[],"_revision_number":1},{"id":"f60955d0c724243e973eddede87697c2bf7795f9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-12 19:02:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84a8dad50980981546630c57e176c3e0d3a4a819","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 19:22:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/49a861b468b749e080e3953e31f4e3f4 : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"012048cec7412b9e74e5ddda7e2450c2d12563a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 20:29:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24275623b06844e5bbb6de2c34ac58b8 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14e3e218e174d1696862c1d9f927c8c : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a176eb70fb224e1c8499bf603e7a0f61 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ab743762ce94beb853fe8eced97884a : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0241f701b1464490b43391c7755204b5 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fb7ad37959b0420682807efe5ec9187c : SUCCESS in 6m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72192b54d4cf4ed5b7b340d129472204 : SUCCESS in 5m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f6aac4a49988498aacb0b55966dff4b4 : SUCCESS in 9m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/728110e56d0642c6a0bd84b17a679037 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/d7a8a906486a4d3d9e4f94b067a9f3f2 : SUCCESS in 9m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/96fd5d6eb47e42c188f87bd2b8a32114 : SUCCESS in 6m 15s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/90573534bc064ce9b0aa7c39da7283c6 : SUCCESS in 7m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cb1678f6e05462bb567802fd033c05a : SUCCESS in 8m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/90e9a32bb0374e9aa5a7fd9a8443fbe9 : SUCCESS in 12m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/6497dfd9baad49c584fe955d67544d4c : SUCCESS in 13m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/5a4168b8c94649fc9dbb8e655f753957 : SUCCESS in 5m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3c91ca1ce4ef42ae89e60d49f3b7c935 : SUCCESS in 24m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/479b33a8dc1a4e4088bdd94a5d13e713 : SUCCESS in 10m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9361ffcca76c4cf6a9d3c234d88ce39b : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de45d3500823441eb06e9c0bd7c749e1 : SUCCESS in 18m 17s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e2e4ca2a841643bc819ed1523362a9d3 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9a001c6d68fc494f81c0b4562190cb9e : SUCCESS in 8m 37s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83c6df12a5c1421e8db03878fbba28d0 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40cec4e1686c4b73afcf7d338cdd1459 : SUCCESS in 45m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/996e918611e44c5796ef52b041865fed : SUCCESS in 1h 14m 54s","accounts_in_message":[],"_revision_number":1},{"id":"8cd3c18572bfceafb0eab8d1690534cedeb40169","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-12 20:32:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e248f56a04757fa3ac6b814a0bfc1b2a99b9b1eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 20:33:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8c8111c8e5a65f041b12f16c5927a9424bb09356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 22:01:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff9c6a8f60d74167836d27bf0edc7bd8 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8bd59438674c1385c7ea0d5de6cb6a : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57ccb76a1194436b8357f2462089f1d6 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a2e0343915f421e80d2b49eb4ecdb8e : SUCCESS in 6m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0644db4270194df6b4935d49c19de28d : SUCCESS in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7516157a055e449dacca9131f7bb6c1b : SUCCESS in 5m 06s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c765294297404a35a3fec4f3ab3b67eb : SUCCESS in 9m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e6a4ed9573f4803a1cd93c4ffe1dbf2 : SUCCESS in 8m 19s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/6debc31c922a4a778ef9aed4dc330518 : SUCCESS in 11m 34s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/103dc9ddf9d242628e879de075db63e6 : SUCCESS in 6m 17s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/215e33fcbe1849fb9d928c5a70d932e9 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c06e1a04aa904882be6cbbbc65fac899 : SUCCESS in 8m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b08dc4b2f38942e2aa474b2ffe98f0bf : SUCCESS in 9m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/aa97280335a8401e8119f668aded0478 : SUCCESS in 14m 25s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/f775cec240534fbf8af6d307cc8a3ca8 : SUCCESS in 5m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bbd82e16c06e479683fa6f74fd40481d : SUCCESS in 25m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/12c6b6ceef4348be97ee6fa32001ce5b : SUCCESS in 16m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3861ba92d91d4761b692b8988d18da53 : SUCCESS in 5m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f778d43ae0ef46e6995a0b55965d52f9 : SUCCESS in 17m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/db3e9beaa3794a26a62562fa42baba7d : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbe4db10bdc74058b5a3c7c7492023be : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0b0a31591f3247df853200b7c0cf3644 : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88d43b3041d64ad7880d03598d574be1 : SUCCESS in 40m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b93f0446da5841c79f7cc98d2aa4e20f : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"b5647048eefe803e64c5b4242613cdd2b4313dce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 22:01:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"239eff5ef04dabf9b08b77a362504fe2395aac75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 22:03:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab847fd5177e43e095c57919e4b1d90c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6b5bf74cd0464e40ea29b0b3de5b2a3479ad6efd","revisions":{"6b5bf74cd0464e40ea29b0b3de5b2a3479ad6efd":{"kind":"REWORK","_number":1,"created":"2021-05-12 16:55:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/791029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/791029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/791029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/791029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/791029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/791029/1"}}},"commit":{"parents":[{"commit":"ce19462764940a4ce99dae4ac2ec7a004c68e9a4","subject":"Merge \"Add python-linstor to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce19462764940a4ce99dae4ac2ec7a004c68e9a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:55:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:55:08.000000000","tz":0},"subject":"update constraint for oslo.db to new release 9.0.0","message":"update constraint for oslo.db to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iaaeeb0b029db04958eaa3c5399ae1580a8ea2914\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ic72017165fbc7fb47b2e7ca53436bc91539c2fde\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b5bf74cd0464e40ea29b0b3de5b2a3479ad6efd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6b5bf74cd0464e40ea29b0b3de5b2a3479ad6efd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
