)]}'
{"id":"openstack%2Frequirements~791031","triplet_id":"openstack%2Frequirements~master~I5b3be6af4f31e0f27b08f5bec71730c6d2251053","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5b3be6af4f31e0f27b08f5bec71730c6d2251053","subject":"update constraint for oslo.config to new release 8.7.0","status":"MERGED","created":"2021-05-12 16:56:22.000000000","updated":"2021-05-12 21:29:23.000000000","submitted":"2021-05-12 21:29:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791031-new-release","meta_rev_id":"2f245f9117b63ca494b69492f3af6463da97b098","_number":791031,"virtual_id_number":791031,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-12 21:29:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-12 19:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-12 19:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-12 17:11:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-12 18:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-12 19:02:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"02d423a4a404ce389703588279f9cef13bbf640a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-12 16:56:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ea1ecd28f2b51f6242a2e425e27ff76e10cc43f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 17:11:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7a3c133cd7134f3b8a0a8eb1e40756d8 : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":1},{"id":"026e040e2cde127cfab2ef012181a42c2ea41c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 18:22:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572a847dbeb14c7caa4454ab8dd917ce : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e62b22beac54448bba699d994904f4a7 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45294b7231c549fd9c3360aa7c5d05df : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fc75e9534614deab460a70aaf927fa8 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6084feab96ef48399c7ce7a2fb49381b : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/296853540b8944089fc4dc37e1f121f4 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33ace2bab0764a12b0a8b74b152a209b : SUCCESS in 4m 22s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a89a1ad2dfcb4afca4b65efda7b1b38e : SUCCESS in 10m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c9c4fe9d9564031bc18700950ee35b6 : SUCCESS in 8m 34s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/aad3e2ec4ca4422eaac26c3ee4213800 : SUCCESS in 9m 27s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/da0353e7306f4c1db04726a38a299fed : SUCCESS in 5m 45s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6de3870aef814d7e95ae282f04cb9090 : SUCCESS in 11m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a5ae4af8251491785860931e30ea0d1 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6e21add7d9b84ad3bdb120ab7b895eed : SUCCESS in 16m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c6d928709e2b4079aa7f2ec6832c1ac7 : SUCCESS in 13m 42s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/055555e7171342daaa94b597146198f2 : SUCCESS in 5m 28s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f492743b6c664ab1bed38c1719e6bf3a : SUCCESS in 28m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/aaf471b46a6947f7a60ffb7c71ffd97a : SUCCESS in 10m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/baa1ca6464e945438766247768b27de4 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1759e3df43684081846b3429e3f3bb96 : SUCCESS in 20m 53s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/1f24228f8c564c46ad3688270e1fc2b2 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5818909aba1b421dac64ffcc185d1980 : SUCCESS in 7m 53s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/82a678d220d148fcb7f6e32549b2b669 : SUCCESS in 10m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5621e29d41df4585ab7a4806325ebe33 : SUCCESS in 46m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37477d4f28fa44c499b6b101d4f10fb7 : SUCCESS in 1h 24m 48s","accounts_in_message":[],"_revision_number":1},{"id":"8c58aac70567ec5ee728a63886745582e0b25838","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-12 19:02:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eeacade6f8a38cc369d94f10947277221899a05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 19:03:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aec35e7fe0a8c2f97858f8cc2442f5bba83f8a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 21:29:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/056d0fcc34f34f59ac1aeb84bf801654 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b18d1118deaf4408ac7e088176110f8c : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ec1d04cf2524d06b4e75813d516d28d : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d60ca787306443f6934d5742e7469759 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1075bc554dfc4f838745e3850a98e6c3 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aff4761cf77e4775924bdceeddc0f4dc : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3798daa5c88640519ab665f6c02f1144 : SUCCESS in 9m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f9edbf7d9cb340ddb18273e59a84b915 : SUCCESS in 8m 53s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/61617e9d75c94b34ba9052fd7fe3b84a : SUCCESS in 9m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/922c139c656249dc96cf3bc4b5202dd8 : SUCCESS in 6m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/1a3cf230fddf46799ca3f31477a8541a : SUCCESS in 8m 50s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/01428f89c64f420089bdbf5a00703f59 : SUCCESS in 8m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0b8ca39986ae453ebb335ca1c2a6a0f9 : SUCCESS in 10m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a0d0742ab0974d2b90cb5e1c60ca2e02 : SUCCESS in 12m 36s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/324f0f025ccf48a29e1744806dbb558f : SUCCESS in 6m 04s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/14f68f86f48f446f8b1f688bfba1d602 : SUCCESS in 26m 49s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/27409ca5c19c42beb805915c0f2523ab : SUCCESS in 15m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/29eef8e397f44197945393ce0a14db2b : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/28c817b3cf584619a4915d11937037b5 : SUCCESS in 19m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5841ef77cbec49149a07d9e0133533b9 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e9a12d88db2c445c90fa979c7344eae6 : SUCCESS in 6m 58s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8a62971c0e3b4b02997a9b9d8587c220 : SUCCESS in 13m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c19dfb075a5c46ec9d73ae01e8b38eec : SUCCESS in 46m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1162e549e74401787934f08794f66a2 : SUCCESS in 1h 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2f245f9117b63ca494b69492f3af6463da97b098","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-12 21:29:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5e9e35c02c1f4bcb5e4e2c126e9f403e4bf2d5b","revisions":{"e5e9e35c02c1f4bcb5e4e2c126e9f403e4bf2d5b":{"kind":"REWORK","_number":1,"created":"2021-05-12 16:56:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/31/791031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/31/791031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/791031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/791031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/31/791031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/31/791031/1"}}},"commit":{"parents":[{"commit":"ce19462764940a4ce99dae4ac2ec7a004c68e9a4","subject":"Merge \"Add python-linstor to global-requirements.txt\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce19462764940a4ce99dae4ac2ec7a004c68e9a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:56:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-12 16:56:19.000000000","tz":0},"subject":"update constraint for oslo.config to new release 8.7.0","message":"update constraint for oslo.config to new release 8.7.0\n\nmeta: version: 8.7.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Iaaeeb0b029db04958eaa3c5399ae1580a8ea2914\nmeta: release:Code-Review+2: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I5b3be6af4f31e0f27b08f5bec71730c6d2251053\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5e9e35c02c1f4bcb5e4e2c126e9f403e4bf2d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e5e9e35c02c1f4bcb5e4e2c126e9f403e4bf2d5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
