)]}'
{"id":"openstack%2Frequirements~791422","triplet_id":"openstack%2Frequirements~master~Ia439548c510da53019018e605b18ad234500b4bd","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia439548c510da53019018e605b18ad234500b4bd","subject":"update constraint for oslo.log to new release 4.5.0","status":"MERGED","created":"2021-05-14 09:00:23.000000000","updated":"2021-05-14 18:06:09.000000000","submitted":"2021-05-14 18:03:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791422-new-release","meta_rev_id":"b69d125afc5fadf4f7d349fa9e75571264088562","_number":791422,"virtual_id_number":791422,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-14 18:03:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-14 14:33:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-14 14:33:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-14 09:19:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-14 10:10:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-14 14:33:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cf4d53545f46f789689cfe8adbae77fcee910212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-14 09:00:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b10a0e06c30a46e4f1433e96ec2106f6c44bcd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 09:19:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3669bc1c4ee14acbaf03e3313442b953 : FAILURE in 14m 28s","accounts_in_message":[],"_revision_number":1},{"id":"0ce49a11c0674cc129ede5613eb00eaa1b41f520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 10:10:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99892c32f71940389c17be12afe0eba1 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449d674ae8eb4b8cb09004123b325aeb : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4091b86108e94348b1551bd80800cdfc : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c7e134c120444f58b2efcc8cd9ed76f : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0b8f65913924101955a0c917063195a : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8c0fff23d594e029552a30745dc3386 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/12ff1b994f784a75a472cb244d0ec4ce : SUCCESS in 4m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ace6d7c6e1a44080ba234f5ac5ebb43a : SUCCESS in 8m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b471c358b9374d1498d66b0e668fe6d5 : SUCCESS in 8m 50s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9fe92e49af944c21842f68928667551a : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/0c736e0dc5d340e7866b8ee8644acb69 : SUCCESS in 6m 42s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/10fdb1deda36430cade42d519eee17c8 : SUCCESS in 9m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7eca91999b634b418d8ffa823a03022f : SUCCESS in 7m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/33fba9c81d094e7db3c33e60932f7968 : SUCCESS in 9m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/466f9a88d6184f6c9d13984200aceb73 : SUCCESS in 12m 16s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/13ea58adb05547f7a9b9dc209f8cdedb : SUCCESS in 5m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/05e948e9d73d4edd878970024666abed : SUCCESS in 29m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d27f679201504946b6c18066fbcd7144 : SUCCESS in 13m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b2eb9ca7ed84ae1b26b89c6b087a70d : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9da7c0c6ea274f509f79ddddc51e3fef : SUCCESS in 19m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/88883056e7704abb86369faebf9cb2dc : SUCCESS in 6m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d84c273f81df4fcdbcc6cd3b0c245f1d : SUCCESS in 7m 41s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5213f5b286fa49ef8cb060ca9554cb96 : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14905587daed42afb3d1058648356dfa : SUCCESS in 50m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a371f1fe825f4e319335b24893998487 : SUCCESS in 1h 09m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f780454396c81d9f2a012bb4f8575a2b1252c49b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-14 14:33:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a38eeba45753426bf13dc6e3397922c264cca6a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 14:34:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3152c81e0381c20e7f1b83965594d546a5b2b8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:03:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873edc14ba7842a095def445f5dc7e58 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d784e9004642bb932c4ef2161d1567 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/966d919eb647465eb9084a4dd21beb00 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b0f816f5aa41fab93fb15e4e0080d1 : SUCCESS in 5m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/34d5775602964e2f8a107cc7991a27fa : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78aee55dd91c473998c227861196ac94 : SUCCESS in 4m 25s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d1d2ebb94a0e4a77b60371f937791793 : SUCCESS in 10m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5c22bc47e83428eb4e8b7e3f54bdb4c : SUCCESS in 10m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/dc25e6fa842041dc885584d2467245b8 : SUCCESS in 10m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/ea75ca7f85d548ef98edd5977abf1d1b : SUCCESS in 6m 37s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f43180d41eed4d999f909b075b07e2c1 : SUCCESS in 13m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6436d213fba44bcfbae9af3b4ef04115 : SUCCESS in 8m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/989c7763283e4385a7eafd30d265cfb9 : SUCCESS in 10m 45s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/b00eaeb5752849eab731962de5ef78ec : SUCCESS in 13m 40s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c52ef43e410d452999c9c3bdb6790660 : SUCCESS in 6m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/98f3f5900e78408186016a69a6e861f6 : SUCCESS in 28m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/968356adf09748bb846dc598ee14d894 : SUCCESS in 13m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8ea580887f384d718992e01ea16c12e1 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45763830b4ba4bf0a82905716239401a : SUCCESS in 19m 22s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/8829898b394b446fb2690e5a24263868 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f6ea65d29b1043e8abd574ccb7009b74 : SUCCESS in 7m 22s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5fa44dd9de444f4da322d1a5d997e4ca : SUCCESS in 14m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3e6bc7896d74b5497ba5d4c46b86647 : SUCCESS in 52m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f82d934c1ab4d0a9b0c5e1822e24069 : SUCCESS in 1h 29m 43s","accounts_in_message":[],"_revision_number":1},{"id":"73bfa1c8451e9afd671440d37dc7a0680022c114","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:03:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b69d125afc5fadf4f7d349fa9e75571264088562","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-14 18:06:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dee6bc5b4444cd69bf040113268a6dd : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"066070d63eb01ea76e52875281fc0720f8897db7","revisions":{"066070d63eb01ea76e52875281fc0720f8897db7":{"kind":"REWORK","_number":1,"created":"2021-05-14 09:00:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/22/791422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/791422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/791422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/791422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/791422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/791422/1"}}},"commit":{"parents":[{"commit":"f00ad51c3f4de9d956605d81db4ce34fa9a3ba1c","subject":"Merge \"update constraint for os-apply-config to new release 13.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f00ad51c3f4de9d956605d81db4ce34fa9a3ba1c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-14 09:00:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-14 09:00:20.000000000","tz":0},"subject":"update constraint for oslo.log to new release 4.5.0","message":"update constraint for oslo.log to new release 4.5.0\n\nmeta: version: 4.5.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Id4b41184be0ec1ee1b6c197753bf0a3c1bfcf706\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ia439548c510da53019018e605b18ad234500b4bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/066070d63eb01ea76e52875281fc0720f8897db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/066070d63eb01ea76e52875281fc0720f8897db7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
