)]}'
{"id":"openstack%2Frequirements~791708","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I4a5fb9678aa52b84dabb9aeb183069cca77e2007","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I4a5fb9678aa52b84dabb9aeb183069cca77e2007","subject":"update constraint for horizon to new release 16.2.2","status":"MERGED","created":"2021-05-17 09:47:47.000000000","updated":"2021-05-17 23:37:24.000000000","submitted":"2021-05-17 23:37:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791708-new-release","meta_rev_id":"e3d3d61aa214d7cbca0d86a6ae3dda89fd06fffd","_number":791708,"virtual_id_number":791708,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-17 23:37:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-17 22:27:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-17 09:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-17 22:27:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 09:48:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-17 11:08:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-17 22:27:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bca4cc9b6cbf9c0fa8d1e6fdb509cc35d5203288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-17 09:47:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca574b96cff567cd055506f8e7de7f96b8944b6a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-17 09:48:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b354826429e2218b853b3a000fd05af51d62c7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 11:08:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/116d725eec6548bab21fbf9045799b46 : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc095dfbb48b45a9ab68e371ebca5d6d : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e85d55c72704bf097bb5201a208a879 : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/42394a65816045a0bb88d8cdbae1fbb6 : SUCCESS in 5m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2cc61146d6884f4f870edd1ea16905a3 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e03f35a2fae64c3aa82787ffbc0d16b1 : SUCCESS in 4m 38s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/66678699d3854d678036a00e48920ea7 : SUCCESS in 9m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/16e22da23daf4a5fbf2516b35450e836 : SUCCESS in 11m 46s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/6c21ce19dc91432daf104b10cc4aa7dd : SUCCESS in 11m 02s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/ff7177487f754949847d2c76661aa1de : SUCCESS in 10m 07s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/7b0a195f3e9240c1a8297a64a546d263 : SUCCESS in 5m 53s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/328effecfc2c437b83111471ba2fa2bf : SUCCESS in 6m 00s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/093216a2e6a347fd95d3772b501a4b0e : SUCCESS in 12m 14s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/57178ef7ab584b2986fde576d26a7306 : SUCCESS in 10m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5fa4ebb888e349b594457007011927ca : SUCCESS in 9m 12s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/4d6c161ba5a145d78522b1b833cc26e1 : SUCCESS in 14m 23s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/cba85a9df1d248718402d8c4e3cb0500 : SUCCESS in 20m 07s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/e3862b93959540cb860eb379586dc33b : SUCCESS in 16m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bf2ecc390d0340eebec931a61a03f3c5 : SUCCESS in 27m 12s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/3a9e4b5a23354c0c810207214bcb98f9 : SUCCESS in 14m 39s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/129028d81a304c93bed9e2f0e9b523db : SUCCESS in 13m 18s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/6768262a53be42ee8467e0c14eda0962 : SUCCESS in 14m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/20121cbc1e2142edb892b9329a145a1d : SUCCESS in 18m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bc50cd21d3b1437ca8d8a546fe197ac7 : SUCCESS in 6m 32s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/f51ba82d5402488e91f94527f7fc4c6d : SUCCESS in 15m 02s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/f22aa42963a64b96aa3e1c56dd39efb0 : SUCCESS in 10m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/208de5ed6b7c4dc88a9a52c6c8db3648 : SUCCESS in 58m 07s\n- tempest-full https://zuul.opendev.org/t/openstack/build/a0635f91ef0144638078b6d2771c7af9 : SUCCESS in 1h 19m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/920c3d15a9a24c928b26d4a113acf965 : SUCCESS in 49m 40s","accounts_in_message":[],"_revision_number":1},{"id":"6b127dbd62d7145ff93746277678d0416416487e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-17 22:27:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b385b7e442543b78ff4b4230370f92821edb6b8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 22:28:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5572e86e030b487aab496b78d7252d2180cf816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 23:37:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/709b72cb89304513a3dcd39c70777c0e : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/07a50f6cf2bc41fab2ea7fb26a09835d : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfcbdd5899c244a7a43164cac33be7dc : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd809a4c146e4f11a67f1b02959f48da : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d0ea4d04c80f4afa815ea113334d0b59 : SUCCESS in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c57e766042a345c48f21e483fe4b2ecb : SUCCESS in 4m 15s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/94071a467d764582a628d811f1f53909 : SUCCESS in 9m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7bbc3601ae28434bb34c3e6d26fd2286 : SUCCESS in 9m 07s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/62b638e9cc4a49699e67248798d09498 : SUCCESS in 10m 17s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/1927af04812b4247838c30914b945abd : SUCCESS in 9m 17s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/586b601cd7fc4560bd50f8462a9f66b5 : SUCCESS in 6m 17s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5f2e99a4884b4ac99d19ba544f61e3cc : SUCCESS in 5m 09s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/a16b3f52a568422a87201851963273b5 : SUCCESS in 11m 41s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3d92fb64fe4f4c70919b76c2de0007b3 : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7767efca3a2541d28fa16af06270a604 : SUCCESS in 7m 58s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/7f30146f9ba74d8898cd94b49a2d200c : SUCCESS in 15m 31s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c4d4f43698914c488ba15241bc44ba7d : SUCCESS in 16m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/167b86a63e6147f8ad99eaee2cbc98fd : SUCCESS in 23m 00s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3fc5261dbd9b429a86677a73182190e5 : SUCCESS in 18m 34s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/6b7be66694d8451c88fbd6a392ca7e40 : SUCCESS in 11m 19s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/b8b1d85a87f04f8ab316dcdc17c7612f : SUCCESS in 13m 29s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/c1071695de2b4c4cb3ab473ac8a39643 : SUCCESS in 14m 30s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d341cf152a641c6ae88f7803e467ef7 : SUCCESS in 19m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/804b784fef3b426f8930b47963ccca41 : SUCCESS in 5m 20s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/d206bba5e6af4f5f87e2ba6be6a86b73 : SUCCESS in 8m 20s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/13fedc258f4f40099b4711e512bc19c0 : SUCCESS in 8m 58s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e996567f25624d06a8cfc37870b6de20 : SUCCESS in 54m 59s\n- tempest-full https://zuul.opendev.org/t/openstack/build/ef774a4a93af49a1a60bdeec6c9f58cd : SUCCESS in 1h 07m 37s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/8f25bbab921d460584c6422c959c6de9 : SUCCESS in 55m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e3d3d61aa214d7cbca0d86a6ae3dda89fd06fffd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 23:37:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44c2f5b79fa999ec0bfaa9377e0de35eb7412408","revisions":{"44c2f5b79fa999ec0bfaa9377e0de35eb7412408":{"kind":"REWORK","_number":1,"created":"2021-05-17 09:47:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/791708/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/791708/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/791708/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/791708/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/791708/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/791708/1"}}},"commit":{"parents":[{"commit":"b169132f0268e12c28486a3cf5eaece0c974024d","subject":"update constraint for os-brick to new release 2.10.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b169132f0268e12c28486a3cf5eaece0c974024d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-17 09:47:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-17 09:47:43.000000000","tz":0},"subject":"update constraint for horizon to new release 16.2.2","message":"update constraint for horizon to new release 16.2.2\n\nmeta: version: 16.2.2\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Commit: manchandavishal \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Change-Id: I57331d57134320c448c9814c20748ee98459652d\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4a5fb9678aa52b84dabb9aeb183069cca77e2007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44c2f5b79fa999ec0bfaa9377e0de35eb7412408"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44c2f5b79fa999ec0bfaa9377e0de35eb7412408"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
