)]}'
{"id":"openstack%2Frequirements~791910","triplet_id":"openstack%2Frequirements~master~I28e610264e1c6768090e9b0e9e80606e5e30fcaa","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I28e610264e1c6768090e9b0e9e80606e5e30fcaa","subject":"update constraint for os-api-ref to new release 2.2.0","status":"MERGED","created":"2021-05-18 09:19:51.000000000","updated":"2021-05-18 22:25:41.000000000","submitted":"2021-05-18 22:24:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791910-new-release","meta_rev_id":"c6c6508f161c4dc4c3cb8d6900d2f8bd644e9260","_number":791910,"virtual_id_number":791910,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 22:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-18 21:00:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-18 09:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-18 21:00:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 09:23:25.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-18 09:39:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-18 10:34:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 21:00:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"959667279abecfbe55467467a432abb504324242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-18 09:19:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2f5027d18fb93f1d7f247eebfbf66829a024c9c5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-18 09:23:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"466d8187acfcc91e127e1b057ba6ea049fdf1e46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 09:39:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e5eb116f613c4d1c8d02ca932c78e891 : FAILURE in 17m 21s","accounts_in_message":[],"_revision_number":1},{"id":"47ff0d1bf84c3a91f910b5e3d490c60c024df050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 10:34:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72968043b868415ba4bc03caa3d8e690 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0651dab101e84bea85a3ccd77635dde0 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c915cb04ee344b79e62f7e0a089c78b : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbb72c4ef1224095b8c0422f5efa9a18 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c3c01c017c5457b90f5ea7d6658b56c : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f27c920d7a7439391b7b7712566f714 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/580f350dbc24407b8ca526b288f235b0 : SUCCESS in 4m 03s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a8f8d895993f4ecaac78988f88fd715e : SUCCESS in 8m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/69657aa9ec7741388acfe827e20b7347 : SUCCESS in 9m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/67b4080b4fe24befac432fbab69da6b6 : SUCCESS in 9m 30s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/01c6b459b6534f5bb9ea33fbc0dce810 : SUCCESS in 6m 04s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/4fe76f6d1d234096852060a8d2051b79 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/63bb8475cac24605baf92c5fa1c796e1 : SUCCESS in 7m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69a1e2a766a849ffa6db087526e4b430 : SUCCESS in 10m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/fd53e6349c0f4967936433f65f01b55f : SUCCESS in 14m 03s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/68a29ef8f13b49059930f311f5b531bb : SUCCESS in 5m 34s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/23071207eb7a45058c9df8da98f2c793 : SUCCESS in 27m 52s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6853bd1235314284ac2bf9455d6650c1 : SUCCESS in 11m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9ca134abe52b4cbe8610d1b0a15d4120 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5ca0acc874b469586c833df2e377ddc : SUCCESS in 16m 38s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/14abbaf1bff34642b4822be8476a4f76 : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b8fff8f9f19a48009a823aef25b421a6 : SUCCESS in 6m 49s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/8ff6ccb8704c4747af53ec3abd1b866e : SUCCESS in 13m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0c32ceb72a5494c86cf4d7ee1901e66 : SUCCESS in 45m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4b8cb05bede640f0939d488ae9059f0e : SUCCESS in 1h 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0ac420c5697a6a65a7acadb10e3f78027c7d877e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-18 21:00:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"850dfd787013ccdd36ff51d1118b29853454f75c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 21:00:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2e840facde12f737c68c6e81908fc59c879aa5dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 22:24:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d9b2ada5e2a46e08533f0bcf8cb7375 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0724af1b44044552963266499af96295 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca168331309b48feae7d3895a503ca6f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90687c97394642d1b1cb079e2e6a094e : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c0f47276bca47b0a04842a3d3c205d7 : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/34ee5fd802534e98ae77447577fcfa44 : SUCCESS in 3m 36s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7d2247a84566422682403b4b9bbdbb52 : SUCCESS in 8m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/161850175f454716bb816fb4b5418209 : SUCCESS in 13m 04s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/88b65d5e91f14cfcba004bf7a287de9b : SUCCESS in 14m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5b34e5d363c2453698fc54e0c9ef85a8 : SUCCESS in 5m 43s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/87abeae5107a41b3965cb3ef91ca2a04 : SUCCESS in 10m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0d9e216b2694e1b94e4062663893200 : SUCCESS in 9m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0bc5cee6556488a9a96c77471900716 : SUCCESS in 10m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a1af149570e64075921037eba6870890 : SUCCESS in 25m 38s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d12db54d83f54e8295174799ea12e9e9 : SUCCESS in 5m 20s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/03decdb096b24870be717cda23ebdb69 : SUCCESS in 24m 42s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/796c763938a64b89a335434559953f76 : SUCCESS in 13m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cbec102731d14028b273a3af72bc83c8 : SUCCESS in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0af9c52fa2bb401f9a2fff700587e21d : SUCCESS in 20m 52s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/97b35d33c38a4c20918df7a98e6a1bf1 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/89b9c0ceab834ea9b526cffc9855ae1f : SUCCESS in 7m 03s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/81158391e0fb4371a0fe0003ee6d5bff : SUCCESS in 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/169bf48d9d4d430089ac0e383b1d1c11 : SUCCESS in 55m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3e4d53fa810d4c2590f3ae80dc8264af : SUCCESS in 1h 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"841a7bb65d24777c8cc9bac4b787cf62d55660c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 22:24:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6c6508f161c4dc4c3cb8d6900d2f8bd644e9260","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 22:25:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca63dbf4fb8476b97a4408cdf6b5c42 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c933e1e601c54388118de4063890c8155e63d6a3","revisions":{"c933e1e601c54388118de4063890c8155e63d6a3":{"kind":"REWORK","_number":1,"created":"2021-05-18 09:19:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/791910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/10/791910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/791910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/791910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/10/791910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/10/791910/1"}}},"commit":{"parents":[{"commit":"62562f5649ad1c49d683904b176a956c4787208a","subject":"Merge \"update constraint for tooz to new release 2.9.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62562f5649ad1c49d683904b176a956c4787208a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-18 09:19:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-18 09:19:49.000000000","tz":0},"subject":"update constraint for os-api-ref to new release 2.2.0","message":"update constraint for os-api-ref to new release 2.2.0\n\nmeta: version: 2.2.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: ricolin \u003cricolin@ricolky.com\u003e\nmeta: release:Commit: ricolin \u003cricolin@ricolky.com\u003e\nmeta: release:Change-Id: I51e58b1795077cbf2cf984dd9c5915a3270bb7ea\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I28e610264e1c6768090e9b0e9e80606e5e30fcaa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c933e1e601c54388118de4063890c8155e63d6a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c933e1e601c54388118de4063890c8155e63d6a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
