)]}'
{"id":"openstack%2Frequirements~792190","triplet_id":"openstack%2Frequirements~master~Idc03994206d4a8fbfe6e7e4f930169993742f8de","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idc03994206d4a8fbfe6e7e4f930169993742f8de","subject":"update constraint for openstacksdk to new release 0.57.0","status":"MERGED","created":"2021-05-19 15:32:50.000000000","updated":"2021-05-20 07:52:07.000000000","submitted":"2021-05-20 07:50:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792190-new-release","meta_rev_id":"2c93e8ab99a3155dad61d16ff3819daba5f789a7","_number":792190,"virtual_id_number":792190,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-20 07:50:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-19 21:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-19 21:34:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 15:50:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-19 17:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-19 21:34:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"fdd010ddbdb55f347622bc529ed5f070ebd8ad19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-19 15:32:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"734f7f8ec151dd1f7d914461630f24ffd8bf67fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 15:50:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c0403d6e7fe74ae58b46d21d3a3c006e : FAILURE in 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"458ff7e74a09db97f25d944c1059fdb6700c5249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 17:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e231ac10f4496aa9df2a6af5859ffb : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9b35aa164714dde9849b6663525b9c8 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/280090a4aca549e1bdf2fc71a9c20611 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8df08fa86a1d4d67a7cd932283d87d84 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d1d6617f8524485968595040cdd0c7d : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8d3a2a7c0a484350afbe6c0ac92273bd : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/209695ed7a3f437494c975529d8aa507 : SUCCESS in 5m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/719ec13bec594cbdb5854fd9f0475ab7 : SUCCESS in 9m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/824cce5f06f94fb59193cc4e8072d97b : SUCCESS in 8m 08s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/366ff38841294d0da22800afc5e1e575 : SUCCESS in 10m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a1c0253e0b7747ea8f24487cfbdc4f01 : SUCCESS in 6m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/db072a162f76449484514bd1777c615a : SUCCESS in 12m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9d5dcd4715484e5f8090d101bd231751 : SUCCESS in 14m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8881a47476914ba7af1f1330e2f82c16 : SUCCESS in 11m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/d5dba0ae36174fe59b43242d0692b149 : SUCCESS in 19m 46s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/704da5cbccc3444aa1e0ae405996ea9d : SUCCESS in 5m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c77e56f92c9d49949cc3e3228e98c267 : SUCCESS in 30m 17s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2a8e8fc8a1c24a859cd18e478e8b950c : SUCCESS in 16m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/54b380a228e34619b83d6198c06cfb24 : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26c15cd09a494c0c80a8e1e7f2e7e004 : SUCCESS in 21m 29s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/eb49d3b778f64ee88aca9c441e64451d : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f32f2a02fcc0484790e6fd04a9039606 : SUCCESS in 11m 08s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b9a830553a945968a4bd2d31b11a0da : SUCCESS in 12m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/029b9127df244cf2a703020725ee035f : SUCCESS in 55m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1973753af4a2464e977433fe2e2730b5 : SUCCESS in 1h 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c94088e21877933b492ead39c5283e31842cab79","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-19 21:34:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7eb13b6325f90f66f0408fc96c37eeb95c73fb5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 21:34:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"120466dd00e9ee57c92f9c9782155e192808c34d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 02:14:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a59a0ce6c7148f98e9f2463dee88a8e : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1214141b37ee454cbb00c4b7c040c9ca : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3727afc779a34083a99cede759fda2d2 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0bdc4d5e21744a78347c1f6202bcb0a : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b7cf46a702f4b8787398c8245c49d70 : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2073c1d20887435aa9f3fa5c97ebf182 : SUCCESS in 4m 07s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2bcb05e939254d75a1f482302652c718 : SUCCESS in 8m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f3abc9bbc022439cbd4bda6583b345ea : SUCCESS in 15m 03s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8bd0c2f07b3747e6ac44ff3d6093639f : SUCCESS in 17m 47s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/cec719039a594d4e85bf534de1b1f208 : SUCCESS in 8m 00s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/d2f6ac68749f4662aa0dea3963ea7856 : SUCCESS in 10m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e263c717d944876a2d0d11ffd4c41ed : SUCCESS in 10m 05s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f779a34b79004170aa92ca9d39763213 : SUCCESS in 12m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/212df2dc9c11498cb1887bc800fead12 : SUCCESS in 25m 19s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9111cc6e9a7e43b4883d9267887dded2 : SUCCESS in 4m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fc36f0651af54f53837584a8e1cc9766 : SUCCESS in 35m 26s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/c84ad7b942a54e73a02600287b8b0e5e : SUCCESS in 21m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9caf06d72e584225a3801d515ce7ed9c : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9469282216ad4fc8bfcb34c199b18018 : SUCCESS in 24m 57s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/57376868fd6d4be78fe6752aa65f5e6f : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67b28f9b0bc74052a745e6f56f81e70f : SUCCESS in 6m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e00ae2b6a9004b51913fb1d580bc7848 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5421e107dfb4426d981844e22ad05467 : SUCCESS in 56m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/be8bb9b4bb0c4c26a3f2a0e1d758c34b : POST_FAILURE in 1h 39m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bcc477b6786b816353ed0ebe9c89455f482ce85b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-20 02:19:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a2e4031d804d6a106aeee396b5072ec94fadde98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 02:35:03.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c8e00ca466544a488065d0886caed4c6 : FAILURE in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"1125cf77f25412652a82d725f5a0d2d3dc602084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 04:36:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca9a7950ae9541eb89a3fa5df05bdd75 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c3720d1ac2041e78e281673587a58da : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7746ee7f6db34db1b7aa82107daac9dc : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e4a1187a9242328ee74f7b63d7b2b6 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddd96487e631421fa88f15009071c903 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a8e48468b8c9401a92c014ae26c223f5 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3059fec1a1944e598f76a0715080359 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e340ce35dbdb4552b97220551c4dc40b : SUCCESS in 8m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b2557cd9a1d643418e79abd0a3e95b7c : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/cad12ee5d81f48cbb4b336231ea4b787 : SUCCESS in 8m 46s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/bf93618818c9493180ec33a5077890f6 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/965d7b9eb4fd42baba8bb03a72406dee : SUCCESS in 9m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ccdc41a5695941be886ac07123ca92fe : SUCCESS in 13m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3c0c243aef974542827ca8361cbe8349 : SUCCESS in 10m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/daf680a35d8047da8b338ffb952ece0e : SUCCESS in 12m 52s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/60a9fb1b27174a0b931f5b23f4c0e350 : SUCCESS in 5m 29s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/bb4ac154e7444f4ca63e42d74ac2215c : SUCCESS in 31m 02s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2f618859d72d42b190ab6638fb09899a : SUCCESS in 11m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d89967a693164113a9cab9d05a031dd1 : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/94673bee11db44919703e3590ff21889 : SUCCESS in 16m 50s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/87a1aa67e5474b6c9af6c5a0e08ebe54 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1866d52fc9b54153a7340603b62ca470 : SUCCESS in 7m 14s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/de96ed7b574e454eab90aba191e7472f : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba08925d78ad4df6bc0ec46719c276a2 : SUCCESS in 36m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0fbd8307f45a450fb9c47e04eee7208b : POST_FAILURE in 2h 04m 49s","accounts_in_message":[],"_revision_number":1},{"id":"9e3c1a57c17dee958128d0d8e32127672e7accc1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-20 04:45:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7cca9596d90be35ce7314cfbee4909bda1ed742d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 05:00:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/95ef6f2deae94141b0cef790e878b420 : FAILURE in 13m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f6e33b722cbba974470302fad93bd6584277183c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 06:31:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877ab503b0be4148a3c058289e26a208 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/436f2230939543b2b133de74bc0c8a56 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ab4a2abd6404d2297a0d230820142ad : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad158b131c8242b0af73ae6076572482 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3304f9ce991a4fbe8a2d0123140f6a8f : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95904523273543e18d49fae084a2a00a : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2bae510768f64f2f8d2ad311431020a1 : SUCCESS in 8m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d433f5e6678b44958d7f03f644693412 : SUCCESS in 9m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f88a91a679cb4cb3921ae9b9e95f5dd0 : SUCCESS in 10m 27s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/2098b88feeae4e0e806d9332be365e44 : SUCCESS in 11m 15s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/306f20f02a4b45978d658435483be18e : SUCCESS in 6m 24s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/2d90b8dcbd294c7990b716918c36a0df : SUCCESS in 10m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/915ffb0f55d240d49bcb2a2a463423a9 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a323e7af144e4ff789866c3192f3d966 : SUCCESS in 11m 10s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/64ab85304180420697db84924087d923 : SUCCESS in 15m 51s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ac02ebd190a64bf58ecb588cc0bb09fe : SUCCESS in 5m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/8eb2745913514e8586a860b5b7deb2cd : SUCCESS in 19m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/98eded5e70db4f248617235c1ef640b3 : SUCCESS in 14m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b8bdb70b7a4a4bc2b2186b6797c8bb0c : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f309ef73b9b549b89086c400892db441 : SUCCESS in 14m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/09319f26ac3543a3bcbf38cef28e170d : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ebd494147189489eb9facd21ae00b82f : SUCCESS in 7m 43s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/b610fb5ae39449f7a2146f09cc7ddb44 : SUCCESS in 11m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aac8d6fca40645cda59a4329358b5966 : SUCCESS in 44m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/16f05615060547e09afd1f07f1f619d5 : SUCCESS in 1h 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"14e94dfa911265c8cee2d6dea2e8aaf344a96038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 06:31:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c369cad9c73723498b777b1c877da617c4af9ae9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:50:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a80bd7780c514fb9862138604b71c7b4 : SUCCESS in 9m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df8eefb62cb40c6ba89677ece427c79 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d69061d911714a7f9408ce06017b5408 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9c60a2377d04036b6965432872ede59 : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/30c58e76544b4d7eba526bff698cb057 : SUCCESS in 8m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0081c17dc162432c9536d587f851c0fe : SUCCESS in 8m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b69d2202f3e84e38be1269ad4e84c1e4 : SUCCESS in 11m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44c1cb13b3514cbda23a46189579af59 : SUCCESS in 11m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/40411458ef47497ea8c05f8e28c17195 : SUCCESS in 14m 25s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/5a95a0385a5740bc99857238118302b4 : SUCCESS in 7m 48s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/245aad9c455a459882bb7fc7215823b4 : SUCCESS in 12m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52b8be2eed2d499d9002d3c48963ad17 : SUCCESS in 10m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c767a650d7a249feabf8ef0246127577 : SUCCESS in 13m 15s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/602a4b4504f34900af3a89ad2ecdc094 : SUCCESS in 24m 26s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/a2d01a07c95c4f398aa0c33ef08c47d3 : SUCCESS in 8m 15s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c977512ab1d74efb95fca841de8a34e6 : SUCCESS in 29m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/00012ba8e32f46f69b5f058b1a732779 : SUCCESS in 13m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2649fd3dd5c407dbb0298277d05b9cd : SUCCESS in 6m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02718a74db1e4322ab07941147ca4c0d : SUCCESS in 18m 48s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/a90cf983cf5b4027a2018a78fe53bdda : SUCCESS in 7m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/daa7a560806c4e9fbdec3552fa6f4840 : SUCCESS in 9m 11s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1ec36e98581b40cda464165dbf6c1cbb : SUCCESS in 17m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58ce92b8c1e04bd6b77a298fffd72c4d : SUCCESS in 50m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/639fd148ecd348dfbbb5085de6d26e37 : SUCCESS in 1h 17m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1aafe4f15fea4f1ef1e5b73db70e160281dbd33b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:50:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c93e8ab99a3155dad61d16ff3819daba5f789a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 07:52:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a519452284d044b2a9880b42c317f890 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"31d3485952a98d3c6d927a3e1edee0052962278e","revisions":{"31d3485952a98d3c6d927a3e1edee0052962278e":{"kind":"REWORK","_number":1,"created":"2021-05-19 15:32:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/90/792190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/90/792190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/792190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/792190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/90/792190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/90/792190/1"}}},"commit":{"parents":[{"commit":"7e351295dfdb204952f8c24d20cbbca35b9c7664","subject":"Merge \"update constraint for sushy to new release 3.8.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7e351295dfdb204952f8c24d20cbbca35b9c7664"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-19 15:32:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-19 15:32:45.000000000","tz":0},"subject":"update constraint for openstacksdk to new release 0.57.0","message":"update constraint for openstacksdk to new release 0.57.0\n\nmeta: version: 0.57.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Artem Goncharov \u003cArtem.goncharov@gmail.com\u003e\nmeta: release:Commit: Artem Goncharov \u003cArtem.goncharov@gmail.com\u003e\nmeta: release:Change-Id: Id95dc32e2cd72f8e3d217146d02160a6cff05485\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Idc03994206d4a8fbfe6e7e4f930169993742f8de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31d3485952a98d3c6d927a3e1edee0052962278e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/31d3485952a98d3c6d927a3e1edee0052962278e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
