)]}'
{"id":"openstack%2Frequirements~792339","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~If0c1e49f8df2fb4c1ad03c65182dd900504616f1","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"If0c1e49f8df2fb4c1ad03c65182dd900504616f1","subject":"update constraint for tooz to new release 2.8.1","status":"MERGED","created":"2021-05-20 09:50:08.000000000","updated":"2021-05-20 13:43:33.000000000","submitted":"2021-05-20 13:43:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792339-new-release","meta_rev_id":"0192971515555bff08054985992fe5ea0d04625c","_number":792339,"virtual_id_number":792339,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-20 13:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-05-20 11:47:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-05-20 11:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-05-20 11:47:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-20 10:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-20 11:12:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 11:40:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-20 11:47:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"7eeed2c732f099f28d8977b04d7ebd9cf5b0ca01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-20 09:50:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"729de3f74441309442a41f536946c2d371d66a14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 10:06:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1f8b3bcad1e543d2a03f9fdc5a8f46a4 : FAILURE in 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"344b4eb30cdcfbbeb2c45f574204871c85559692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 11:12:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818ad4e5b1024775b86764db2dfed8e9 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bcd4553629544fc87d5d7475283b33f : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52ced9162ba442ad8a46a09d09655eb2 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5256deaf08fd4afd97d4c7902a937600 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c7567797c955485fa0f9b777f4d17dc4 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1410afd14dd4a91b0719cb8844171d4 : SUCCESS in 5m 37s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2e38634b42474ca1911fbbfa4bc51f1f : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ca4fbf5d1e834736b9fc17b427cc29c9 : SUCCESS in 9m 21s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/eac248b9a6c648c38e882603d58919e3 : SUCCESS in 9m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/c0c26c409f2744a4b7987ba09dd4f3d5 : SUCCESS in 6m 21s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/36f1398479014985b8bd4b336c7bf8c0 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5f695cd2979c4f44bfe55fef98f9c0c5 : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/23472db819e646049b8f8bdaa7bdff3f : SUCCESS in 10m 59s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0e267bfa971d4f38bb21a1ef42439bbb : SUCCESS in 19m 37s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/00b2c066bd264131938913d92b37d749 : SUCCESS in 5m 48s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/d5a15cedfb6e41b3ae7e4c477d1c595f : SUCCESS in 23m 00s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/4661c75874874f41b419990372f87ab0 : SUCCESS in 11m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d79b6cf7dd84943a67e060304a3b42a : SUCCESS in 3m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82da785958ff4611953ae5b9bff52815 : SUCCESS in 17m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/b5d79d35e3da4574a8072305189d5b93 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3d3437d9ff6940db8dd403c1bebad1c1 : SUCCESS in 6m 54s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/876096edf6804ee9b1a4135e07b0ad79 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df153073a7344420971a3a1771665b46 : SUCCESS in 50m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8c3d0900f9e478ba03ba14222703c31 : SUCCESS in 1h 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9d7caed884c9ed75a931a8308c21fc2c5bd69603","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-20 11:40:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d9ca83cb191fb97888d98f0f54f71f0b4cb4007","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-05-20 11:47:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2e67bd0af17031946284ebcb2aa164cb56b61e65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 11:48:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f60585e6661fd3b236799af0b71edbdbaa4061aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:43:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/267287cfd490462396d641a42c8e814c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a45eb063f636418eac9709ad6de983cc : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8e0318d14b647a0a75aa00c9522914e : SUCCESS in 5m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/96983d3314434df992efde8956d9c0c8 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8046a38da139411a99146aba44b723d8 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/964a2df45d614811aad350f7d88ebd9f : SUCCESS in 12m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/189dc32bebdb4f0d86bacd54db2fa2a7 : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/a6b25fb19ff343898adc1e4765338c57 : SUCCESS in 9m 39s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a271fc6e5b7e4b7abdfd21ffabb09058 : SUCCESS in 5m 58s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/39baea2f87394f898299627190b73da2 : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/487a8b0425684ed0b8eff342ebe8d45d : SUCCESS in 9m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15b0414f6dca421c856c51e5a1975a8a : SUCCESS in 10m 34s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a73185b16ea34da48a13e2cdc274cbde : SUCCESS in 14m 49s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/3f8c49cc0e844139a43a5868ad456cf6 : SUCCESS in 6m 30s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/224968ce1e93427dafd9a67de4116446 : SUCCESS in 30m 03s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/70b6b2673673494995d05b094fb7252c : SUCCESS in 12m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ba5dc6322094876a7c9cc7107d754ff : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/87c507f14bde44e68d5d704050fc21dd : SUCCESS in 22m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/92179974ce5848c983854b8854184278 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/239aefc4649546e982b18584f393f678 : SUCCESS in 7m 23s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/aaf68054d0944d50a8bb02bf26454ee6 : SUCCESS in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a49f110016a4af19871e6985950c35e : SUCCESS in 50m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2dd40c0f97e6481398c4ea9820eca79b : SUCCESS in 1h 20m 27s","accounts_in_message":[],"_revision_number":1},{"id":"0192971515555bff08054985992fe5ea0d04625c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 13:43:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"98091329cf49a3abc4840343401711bdb4f05bd3","revisions":{"98091329cf49a3abc4840343401711bdb4f05bd3":{"kind":"REWORK","_number":1,"created":"2021-05-20 09:50:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/39/792339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/39/792339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/792339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/792339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/39/792339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/39/792339/1"}}},"commit":{"parents":[{"commit":"4065d29765562061aef11056036a76fcc1153073","subject":"update constraint for tripleo-common to new release 15.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4065d29765562061aef11056036a76fcc1153073"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-20 09:50:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-20 09:50:07.000000000","tz":0},"subject":"update constraint for tooz to new release 2.8.1","message":"update constraint for tooz to new release 2.8.1\n\nmeta: version: 2.8.1\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I27c938feeea6ee95551bd3d103694dc9e849698e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If0c1e49f8df2fb4c1ad03c65182dd900504616f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98091329cf49a3abc4840343401711bdb4f05bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/98091329cf49a3abc4840343401711bdb4f05bd3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
