)]}'
{"id":"openstack%2Frequirements~792733","triplet_id":"openstack%2Frequirements~master~I857f7be1e1fd7532cfb63318a23e3def3b9f7c1d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I857f7be1e1fd7532cfb63318a23e3def3b9f7c1d","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-05-23 06:37:13.000000000","updated":"2021-05-24 14:59:12.000000000","submitted":"2021-05-24 14:57:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"792733-openstack/requirements/constraints/noclob","meta_rev_id":"b9d3e3ce6955501c55d8a4e48506b1abc5f333b6","_number":792733,"virtual_id_number":792733,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-24 14:57:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-24 13:44:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-24 13:44:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-23 06:42:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-05-23 07:01:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-23 07:05:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ad8cc0af54ba4df880990d285d8bf14c727137a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-23 06:37:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d300824786d02d509f13c08df55eddfca27e2ad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-23 06:42:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b16ca5c568529cf4326ef658c6664f796268897b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-23 06:43:10.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"9383fdc3606031eb96ae3035a5ff94c4a6f5b535","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 07:01:58.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8825ab8832ab4a7abcfa071317c4a069 : FAILURE in 16m 49s","accounts_in_message":[],"_revision_number":2},{"id":"d38a9f83cb38ccc555c11a84905ef75678044709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 07:05:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053c2c0661a2499783ce3b8ca7ee7c6e : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9b7311b7e246a59e2ec89a76b92709 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b334a79c4dc442d29b037f203824aa66 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26a1ae3cedbd4582bbb8d17994dce8a3 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d79567f9442d47c984edb6c9851a89af : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76c7e3d695a8466ca657548777e71a8d : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfb7b601bd0e4ff99fb6d3b02220edfe : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cb164ee37fca4704811b8efa260e8e9b : SUCCESS in 12m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/14e0c7b0e85041d3a959c30ef7cdb015 : SUCCESS in 10m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4248d34c295a4fad891991c55f549987 : SUCCESS in 13m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9154a3b9a1574f2695dabdce1ce66b1c : SUCCESS in 6m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2fa3551bd24247a4a39c82069ca5db6f : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29d3b185c46f4e569e13aa5c399fe90d : SUCCESS in 7m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49bfde80d561492891197458f8ffe9fe : SUCCESS in 12m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2418f37992ca4057a70735e2f3b838b6 : FAILURE in 15m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0c40d4d06a8f4462ae03bfb165828e1b : SUCCESS in 9m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/82159d5540b7459baa55ec76481540d1 : SUCCESS in 5m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4bcea34f78f343cd8f7a9a5efc9b8721 : FAILURE in 6m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b564ccda9c354e97b7f374a01cebbb60 : SUCCESS in 13m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2bdfddc3fe874dfe9460134c8736db95 : FAILURE in 4m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8b2cfc36a023425fad8c9dd6404b0d53 : SUCCESS in 14m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c39a4a5225b84776bea6c2dcec4af542 : FAILURE in 4m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d14e1242030a42d88b3f1bb9d33e0d3d : FAILURE in 5m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7f19a1405e984633af2c98b572851a2b : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b931430b0ff4e84a3b92b44cdcef830 : POST_FAILURE in 21m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efb4a19d467d41d284c99ec1481b4161 : FAILURE in 19m 48s","accounts_in_message":[],"_revision_number":2},{"id":"35b8c8ea1635de2263dea69be819d4f4cdca86f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-24 04:20:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2feaba7c5c533205b4363ac49edc76c9550be011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 04:38:29.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7a599c91e3774bed899abb6fc81cbb66 : FAILURE in 15m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9ff273a73617528a0615652729bd803994c11d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 05:41:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eecf4082039c41b0ab3e769d7dfd75cc : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/935ff583a0c2452a8fd48b89c41f65f3 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0572be4c374447a0be7bd7405051d9ff : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b594692a9f54cefbae61fff8f4f9a56 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03d95dfa3dff4bed94f891bcc86b2403 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ef9e858164e440bb57b0ff87f0f2b7a : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c65540efb174ee795b81ab83973e7bc : SUCCESS in 3m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4a29b6841e9e4156901d5a53f849835e : SUCCESS in 10m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c1a1da3b0e5b4615a6b3f46ff9fcede9 : SUCCESS in 11m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9fa5f12057fb402f99bd12f833b05b0a : FAILURE in 10m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/be16a7392f6c46c8ab5a24877d04fe2a : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/74d1d9d97c0e479f8ca36123eaeda514 : SUCCESS in 11m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b91c24538cf4f7199e2ac1a987ee753 : SUCCESS in 7m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ae29e44ae87b43b991e67a1e061f5cc2 : SUCCESS in 10m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/64f8cc67c75f4799beff29f8e7149923 : FAILURE in 18m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c6b420b021b14eac8980c6a228ad07b9 : SUCCESS in 9m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/36bd56651a5f4feaa85a5a23555a6d5e : SUCCESS in 7m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1fefc600a9d5436a9f12d86bd1d5f33c : SUCCESS in 27m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/16429db14d6447b78f52447faac215b3 : SUCCESS in 13m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2a89d04770db4fa29bc9f39637687f1a : SUCCESS in 3m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d95d9b5d88014671ba82cf43463dfcfb : SUCCESS in 17m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/19d44320d8ac421bb0ad088220f6ed59 : SUCCESS in 5m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b5c881ada9a1426b8536e060c214ea6e : FAILURE in 4m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3b1faa1a630b439ca5fb6c70f5767649 : SUCCESS in 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b142999ea2f24d6185a8ba4f2e85409c : SUCCESS in 49m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4db4a83affe548f7a5dc3e77e9577e7c : SUCCESS in 1h 19m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9549c5877d4836f1a00f73a590fccc49bb7f8c93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-24 06:10:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b9e3d8cb5517ed731ce02108ac92a33308bef64f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 06:35:25.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1b9aaa5478df4e1084838996f3c48921 : FAILURE in 16m 26s","accounts_in_message":[],"_revision_number":4},{"id":"a5b922ef642c9fa778dc935bb5aa7d4357812959","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 07:52:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62fee2c40c594289ae972b62fe130dd3 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b94211f93064a54b1890c0c5371d306 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9e510fd2e3741d5aac4d84696cbba6a : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1816a68d5174f65b944e4e710a5613b : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ed3031836924c9a98bab327121cc112 : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/07de3c7484a1424fa448ef4ef65a98fb : SUCCESS in 7m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4cfef134b6744d37b6fe2b3e84701214 : SUCCESS in 7m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a3dbfde07b34b75ad547258b7e9667b : SUCCESS in 13m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94fec46dd85a4887b3760dd383422f70 : SUCCESS in 12m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/54f95596208f4233853ea62c3d81bb66 : SUCCESS in 13m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/15229690782d45b68c56cbdcf4fe2ed4 : SUCCESS in 9m 25s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4bec9a5ca42d4557b3d2fe9324cab0a0 : SUCCESS in 11m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b34e153c5e944c779ea08c4918d12a26 : SUCCESS in 11m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fa9e061d38a04348afb3a37654d1cfed : SUCCESS in 14m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/223ad13e86694fc3a291e036c964a797 : SUCCESS in 16m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a62cd1cbc8e8479297de9052e64c0f22 : SUCCESS in 12m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6cb18057a6b84b60b5955d535fd91521 : SUCCESS in 7m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/be9faeb63a554a2c8298cd6e6f629e4e : SUCCESS in 29m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/28a0558929f2401b9506f46a42222024 : SUCCESS in 16m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9bbc66245354aa683c6213349cb115a : SUCCESS in 5m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c1db4b48dad54baf986a988d2523fe76 : SUCCESS in 20m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8cec2b937b0b4545831a6db0a7afae0e : SUCCESS in 7m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c1b1a65281641348a986edff3cc9f56 : SUCCESS in 9m 26s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5283822493a644adbbb82f25f5fe7625 : SUCCESS in 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d79c6f8884b45b8869a2647b2e67af4 : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/450b0d8fefc845c79cc182c9b695c422 : SUCCESS in 1h 28m 28s","accounts_in_message":[],"_revision_number":4},{"id":"40097b5b880c68f79e7ca32bdec28e541e1439c1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-24 13:44:38.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"05931e0ad91efc928d9b41dd855b166cef43d0eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 13:44:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"23db3cf9cfff8f06dfa740db4354f6477e384d20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:57:54.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a57eb04e084c4ec78fa3aa8157577ae4 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60fb463123994c05a38243a998fdd890 : SUCCESS in 3m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb7eba98eed146d289df99ae9585cad4 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5e01bf560045a79663d94f64b5e92e : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/65fd6742afcd4b41a12f3c36155a8b0d : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b30885d31a14ec5850830b6a0d9f1c9 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b355b966d32447199413287998b0eb0b : SUCCESS in 10m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/34060208ed8544daae55f1a06b1860bf : SUCCESS in 10m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d9b6f315cc6e41658564c1bbe52c3ed9 : SUCCESS in 10m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/20d60dc351d1478fa278f990fd039403 : SUCCESS in 6m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/805327e4cdf6424593960107bf5de0eb : SUCCESS in 8m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b762df023e541e09472054cb8303d04 : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8ff3365253254fccb12371884686e516 : SUCCESS in 10m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/be280a05484d4ad994621d181011c2c7 : SUCCESS in 14m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bda166909b4e4dac8c4dfe8db0da8d6f : SUCCESS in 9m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cdc9d4a0357e444d9473a22d7c7862fc : SUCCESS in 5m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d7c5c186108b433db10a0f06783b9b65 : SUCCESS in 22m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fac6b40e07a24dbbae9f58f332fc09cf : SUCCESS in 12m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1f95e5c681f4d3b9166a45a8032f9d7 : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/83aabba8608c4ba5a098d5ca05be4a94 : SUCCESS in 19m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1dbc29a6235b493ba392c4d9d69288fa : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49f60ddd6ba442a09d9f3be52fe3ebad : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/70eb2831be1f4dc38ab94289f839a4cd : SUCCESS in 13m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8330850eb4fe4cab82036c982799d781 : SUCCESS in 43m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3f66bd826a347c990d3ae9c7744645a : SUCCESS in 1h 11m 36s","accounts_in_message":[],"_revision_number":4},{"id":"00ceef51758c99b2426433a27227a5a1307e0fee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:57:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"b9d3e3ce6955501c55d8a4e48506b1abc5f333b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 14:59:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c72951ffb74345489b271130bb8263ac : SUCCESS in 56s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bb7ae87daf2bf2b515c73e6375f552df6836f273","revisions":{"705676d9bcb09c7e652fdd46feeefdaab3023206":{"kind":"REWORK","_number":1,"created":"2021-05-23 06:37:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/33/792733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/792733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/792733/1"}}},"commit":{"parents":[{"commit":"70c16cff5c9cc7dd028cd8d3a9056c3965916237","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70c16cff5c9cc7dd028cd8d3a9056c3965916237"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-23 06:37:11.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-23 06:37:11.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I857f7be1e1fd7532cfb63318a23e3def3b9f7c1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/705676d9bcb09c7e652fdd46feeefdaab3023206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/705676d9bcb09c7e652fdd46feeefdaab3023206"}]},"branch":"refs/heads/master"},"eec10edc68328e9d9a1ce9ba8ce6fa7afffa9559":{"kind":"REWORK","_number":2,"created":"2021-05-23 06:42:59.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/33/792733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/792733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/792733/2"}}},"commit":{"parents":[{"commit":"70c16cff5c9cc7dd028cd8d3a9056c3965916237","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70c16cff5c9cc7dd028cd8d3a9056c3965916237"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-23 06:37:11.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-23 06:42:54.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I857f7be1e1fd7532cfb63318a23e3def3b9f7c1d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eec10edc68328e9d9a1ce9ba8ce6fa7afffa9559"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/eec10edc68328e9d9a1ce9ba8ce6fa7afffa9559"}]},"branch":"refs/heads/master"},"a1d1281b637a9d96342ab22b1a1de97ae0c5ca93":{"kind":"REWORK","_number":3,"created":"2021-05-24 04:20:57.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/33/792733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/792733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/792733/3"}}},"commit":{"parents":[{"commit":"70c16cff5c9cc7dd028cd8d3a9056c3965916237","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70c16cff5c9cc7dd028cd8d3a9056c3965916237"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-23 06:37:11.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-24 04:20:40.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nneed to make a review for pyroute2-0.6.1\n\nChange-Id: I857f7be1e1fd7532cfb63318a23e3def3b9f7c1d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1d1281b637a9d96342ab22b1a1de97ae0c5ca93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1d1281b637a9d96342ab22b1a1de97ae0c5ca93"}]},"branch":"refs/heads/master"},"bb7ae87daf2bf2b515c73e6375f552df6836f273":{"kind":"REWORK","_number":4,"created":"2021-05-24 06:10:56.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/33/792733/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/33/792733/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/33/792733/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/33/792733/4"}}},"commit":{"parents":[{"commit":"70c16cff5c9cc7dd028cd8d3a9056c3965916237","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70c16cff5c9cc7dd028cd8d3a9056c3965916237"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-23 06:37:11.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-24 06:10:28.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nneed to make a review for\n\npyroute2-0.6.1\ndocutils / sphinx\n\nChange-Id: I857f7be1e1fd7532cfb63318a23e3def3b9f7c1d\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb7ae87daf2bf2b515c73e6375f552df6836f273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bb7ae87daf2bf2b515c73e6375f552df6836f273"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
