)]}'
{"id":"openstack%2Frequirements~792776","triplet_id":"openstack%2Frequirements~master~I92888fdbf80c1f58fbef7dc8b35404a9ec57ca5b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I92888fdbf80c1f58fbef7dc8b35404a9ec57ca5b","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2021-05-24 06:21:47.000000000","updated":"2021-05-24 15:28:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"25a6227cd061b3586e034336702bf7ea9bf817f3","_number":792776,"virtual_id_number":792776,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-24 06:59:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-24 06:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-24 06:59:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31d281907c7bb9a57fd4c23e54bda8192accf006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-24 06:21:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"811811c5be45d394fdd0e2356095d72206d81ab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 06:37:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/22bf4dec6da943b09a614aac4b5440d3 : FAILURE in 11m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0a44ee34cba874166dec403b14a992375354c663","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 06:59:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b77d347562b44948a0b9863366f15aaf : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51f91a025d564df0b3c7a64404831822 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d46e9d69c0f49ffae6fcecaac0c300e : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ded5081f9c54b4fbdd96608144c0ce6 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80caedd5bdec49a694e143f2e6413a98 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b9d4515829694710963b38debd500507 : SUCCESS in 6m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8fbff816e9524df1919ca6e0edb3fe41 : SUCCESS in 7m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/14d9b2021eff42509edf022a0d1fe446 : FAILURE in 7m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5bdb75695cbc4e9f8fdc6c06b0e8b007 : FAILURE in 7m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d7229b43443f4fc09d51b2ca39748a18 : SUCCESS in 11m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0d0f741a43f0426dae846ac0967d8fc2 : SUCCESS in 8m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5649f55d49094b378e9c38feea961cea : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/40edcf1eda48483c81eec313d8cbacd3 : SUCCESS in 12m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ae2c42b6ad0f431d8516065a8afb300c : SUCCESS in 13m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4ece2a04be0c4c5aa8d978070ceee6bb : FAILURE in 16m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/40f2346cbd9d4ffb9549c2b16a1f597b : SUCCESS in 12m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a5b131c83fe24114964842b8041b914f : SUCCESS in 7m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2ae2496460324b15968a95d6e686ea91 : FAILURE in 8m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/429bce5b26194b4f94ea0bdb9e4b07f2 : SUCCESS in 15m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/eeb3e65d018649eba45a44292b81b13a : FAILURE in 6m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/69296f31f2f44bfcb7c612b35fe24786 : SUCCESS in 21m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/364d2e9d176146ac8883bd69b9969698 : FAILURE in 7m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/adcb47b13b5c47d883051963f0eed332 : FAILURE in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/89b16e32037d418fab8f3ea6fbdf2615 : SUCCESS in 13m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23b78fd7ba1a429b840aabd3114eaecb : POST_FAILURE in 28m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ab3055c5d8a4f7e99588a0a63fe70dd : FAILURE in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"25a6227cd061b3586e034336702bf7ea9bf817f3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-24 15:28:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3eacc21608a2ccef24c164c1f754e83b68475d9f","revisions":{"3eacc21608a2ccef24c164c1f754e83b68475d9f":{"kind":"REWORK","_number":1,"created":"2021-05-24 06:21:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/792776/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/792776/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/792776/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/792776/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/792776/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/792776/1"}}},"commit":{"parents":[{"commit":"7a55d689e00d92de938a5c9077719faf8f980daf","subject":"Replace assertItemsEqual with assertCountEqual","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7a55d689e00d92de938a5c9077719faf8f980daf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-24 06:21:45.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-24 06:21:45.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I92888fdbf80c1f58fbef7dc8b35404a9ec57ca5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3eacc21608a2ccef24c164c1f754e83b68475d9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3eacc21608a2ccef24c164c1f754e83b68475d9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
