)]}'
{"id":"openstack%2Frequirements~793022","triplet_id":"openstack%2Frequirements~master~Iee819f254bed3747f120010c27957d3a14c2b0e8","project":"openstack/requirements","branch":"master","topic":"docutils-sphinx","hashtags":[],"change_id":"Iee819f254bed3747f120010c27957d3a14c2b0e8","subject":"update docutils and sphinx","status":"ABANDONED","created":"2021-05-25 17:58:20.000000000","updated":"2021-08-06 15:43:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"cec36e99a94c250693eec6e3766318b96c046e8d","_number":793022,"virtual_id_number":793022,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-25 19:21:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 18:16:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 19:21:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 08:06:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"CC"},{"updated":"2021-05-28 08:06:51.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"}],"messages":[{"id":"d50bb6dff3c63bf10594c93cc789a2f460158fb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-25 17:58:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5541af099691acba8cd4d856f756793db2e57cfd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 18:16:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a0e719f337fd4136972e11fba70bbb23 : FAILURE in 16m 22s","accounts_in_message":[],"_revision_number":1},{"id":"cca73bc1974be099d7008fd4a9edc126d2e468c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 19:21:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee4d3f0e100b4e678e3bf3e654cecfba : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/629e29d77b8448158ce8ee50328b193d : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f5d8d7d995e4cbaa51ba1827eae2cb2 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a94f6d2525914f45a26ff7a59f76508d : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/878c352a629a491ebb8935daf7bc9430 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/39f6907d1e7942fdb9dc1645cdc486e0 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f0c995d952614ee081ad7e8834ed2374 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8be8bbbec50b479caa9e1cb5a8d3d089 : SUCCESS in 12m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29d593c65a834f8cba5770b40936a44c : SUCCESS in 8m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5f232e91f25244d0a9a10e0f9bc56260 : SUCCESS in 11m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ab655d422aac489da560b98e60c7252c : SUCCESS in 7m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/578f933883554e85acb13fa81e8a9273 : SUCCESS in 10m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f55cc98f5e3c445993040b698b170cfc : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4e3d6f07fa5c4730b81fdb2f9f153f20 : SUCCESS in 14m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8370b04107b54e08a17ebaf9266d6519 : SUCCESS in 11m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/611d9e7e35e74e708af92f4cc22cec54 : SUCCESS in 10m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b9fd4e3bebcf48d08e37b97fdd169a73 : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/17f698f7e7f74753b544a50c493fc49e : SUCCESS in 34m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b578133ab3a64610af0c96830d13f2ca : SUCCESS in 14m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8dbfa617330449a3ba890e175642513b : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9ab2425594cb405481574e644419791b : SUCCESS in 18m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/49e83b2454c14224833379eb35bdea04 : SUCCESS in 6m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/197e7e5105194006a7fbd274fc5734e5 : FAILURE in 6m 05s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c8abf7f11b2a44e1b5fb0dd64741fb80 : SUCCESS in 10m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d66def2c8284c24a00fed3268d2ea0e : SUCCESS in 49m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c63292d4fc746429caf24bca6acc3a9 : SUCCESS in 1h 21m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0b1d85bae60677bd958369fa662ffea795fb0d05","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-28 08:06:51.000000000","message":"Patch Set 1:\n\nLooks like this needs an update of openstackdocstheme:\n\"[openstackdocstheme] using theme from /home/zuul/src/opendev.org/openstack/python-openstackclient/.tox/docs/lib/python3.8/site-packages/openstackdocstheme/theme\n\nExtension error (openstackdocstheme.ext):\nHandler \u003cfunction _builder_inited at 0x7f1df63c8160\u003e for event \u0027builder-inited\u0027 threw an exception (exception: make_link_role() missing 1 required positional argument: \u0027caption\u0027)\"\n\nSo, let\u0027s not merge this for now, it will break a lot for sure.\n\nStephen, could you look at this, please?","accounts_in_message":[],"_revision_number":1},{"id":"c69a6d3f1366cd49e2f768ef185eab3e72c9925c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2021-05-31 15:58:47.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Looks like this needs an update of openstackdocstheme:\n\u003e \"[openstackdocstheme] using theme from /home/zuul/src/opendev.org/openstack/python-openstackclient/.tox/docs/lib/python3.8/site-packages/openstackdocstheme/theme\n\u003e \n\u003e Extension error (openstackdocstheme.ext):\n\u003e Handler \u003cfunction _builder_inited at 0x7f1df63c8160\u003e for event \u0027builder-inited\u0027 threw an exception (exception: make_link_role() missing 1 required positional argument: \u0027caption\u0027)\"\n\u003e \n\u003e So, let\u0027s not merge this for now, it will break a lot for sure.\n\u003e \n\u003e Stephen, could you look at this, please?\n\nI\u0027ve a series of patches at [1]. It\u0027s not complete (I haven\u0027t done a full visual diff of before and after) but it should fix this immediate issue. It does seem like Sphinx broke their own stable policies for this particular error, in that there was no deprecation cycle for the API change, but I guess that ship has sailed.\n\n[1] https://review.opendev.org/q/topic:%22sphinx-4.x%22+(status:open%20OR%20status:merged)","accounts_in_message":[],"_revision_number":1},{"id":"cec36e99a94c250693eec6e3766318b96c046e8d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-08-06 15:43:12.000000000","message":"Abandoned\n\nthis was done","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0719ef6bd5178dc6a0e5e7347466587f71e970f4","revisions":{"0719ef6bd5178dc6a0e5e7347466587f71e970f4":{"kind":"REWORK","_number":1,"created":"2021-05-25 17:58:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/22/793022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/22/793022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/793022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/793022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/22/793022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/22/793022/1"}}},"commit":{"parents":[{"commit":"bc4ca285cb46171c7ca3df13abe1689174e43111","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc4ca285cb46171c7ca3df13abe1689174e43111"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-25 17:58:04.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-25 17:58:04.000000000","tz":-300},"subject":"update docutils and sphinx","message":"update docutils and sphinx\n\nChange-Id: Iee819f254bed3747f120010c27957d3a14c2b0e8\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0719ef6bd5178dc6a0e5e7347466587f71e970f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0719ef6bd5178dc6a0e5e7347466587f71e970f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
