)]}'
{"id":"openstack%2Frequirements~793024","triplet_id":"openstack%2Frequirements~master~I40aa43570b8a19642df434bd43a1a94a5325d131","project":"openstack/requirements","branch":"master","topic":"flask-2.0.1","hashtags":[],"change_id":"I40aa43570b8a19642df434bd43a1a94a5325d131","subject":"update flask to 2.0.1","status":"ABANDONED","created":"2021-05-25 17:59:53.000000000","updated":"2021-05-28 02:07:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ee919b1be9f57e55f4e748c0ff32e6c84055c8b4","_number":793024,"virtual_id_number":793024,"owner":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-05-25 18:34:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-25 18:12:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-25 18:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c0079fdf377f06c4bef0731f508e26e7e48f907","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-25 17:59:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8bdfc7904a6d9c974582cd6fd4ab8bfca5d4899","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 18:12:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c8e7923d3eee4d65b5d98d410b76fa1d : FAILURE in 8m 37s","accounts_in_message":[],"_revision_number":1},{"id":"14f8f35cd48333460e9dad4f5f05291152611b40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 18:34:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45eb29c418b94f63b4422071209bf08a : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb5637fccd814603bc39020404b0c2d1 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7da8e002200b4ffaa27f0bb0cdb4d505 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30133c784d1546a2b13f71013049a69c : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f3cfb0a34a4aea90cb33838c8ec7e7 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b0ca1ffabfc84ac9a56bcc36bfceb9cd : SUCCESS in 3m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/eed6e54ef7274736b430c561b83b1357 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cf26e0f1f3c74ed0be57dcfce2ec6cb9 : FAILURE in 6m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f00d1c9cac404e188a8022631ef59357 : FAILURE in 7m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e72d49c379d24922a0a431b5143b7513 : SUCCESS in 9m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/075385dfe71c41038c57e9670f169eb2 : SUCCESS in 7m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/56add62d69614ba491b5c472c2ac7094 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08d879dd8954452094b648941eebad38 : SUCCESS in 11m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/02a7c2fda69c450ebd07bb2af18bf468 : SUCCESS in 11m 02s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e3f28b8100f642fe9befbf2109201acd : FAILURE in 5m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a9979751f01a4760bbe88f5f1408e519 : SUCCESS in 9m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1fd9d55e7fef4a42aed2f7d35e8ada84 : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f15c8cef642d41199900a5534477afdc : SUCCESS in 29m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fbd5ce6a8017425b9853b89029d8204c : SUCCESS in 11m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e838c86de17b47d989146f7b0faeba82 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/834626a50e4f4f83b2487626ebf4d2f3 : SUCCESS in 20m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7bedf96285746b59e5541bdc4fb1e9d : FAILURE in 4m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/936d2053083f42019a2aa9685509fa0f : SUCCESS in 7m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e8ad71b2e9794ffe997a6dc52ae0c428 : SUCCESS in 12m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/022da335016343f1a9b4fd5a1beb2a91 : POST_FAILURE in 24m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6f2fe0eb8a8a497e9b69c867b07b5c91 : FAILURE in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ee919b1be9f57e55f4e748c0ff32e6c84055c8b4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-28 02:07:39.000000000","message":"Abandoned\n\nrolled into the werkzeug review","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"76a53fb24f388930c4cddc906ab084f336229860","revisions":{"76a53fb24f388930c4cddc906ab084f336229860":{"kind":"REWORK","_number":1,"created":"2021-05-25 17:59:53.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/24/793024/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/24/793024/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/793024/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/793024/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/24/793024/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/24/793024/1"}}},"commit":{"parents":[{"commit":"bc4ca285cb46171c7ca3df13abe1689174e43111","subject":"Merge \"Updated from generate-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc4ca285cb46171c7ca3df13abe1689174e43111"}]}],"author":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-25 17:59:42.000000000","tz":-300},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-05-25 17:59:42.000000000","tz":-300},"subject":"update flask to 2.0.1","message":"update flask to 2.0.1\n\nChange-Id: I40aa43570b8a19642df434bd43a1a94a5325d131\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76a53fb24f388930c4cddc906ab084f336229860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/76a53fb24f388930c4cddc906ab084f336229860"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
