)]}'
{"id":"openstack%2Frequirements~793328","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I912b47ed6a91ad958dbe19242f3930669b603ba4","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I912b47ed6a91ad958dbe19242f3930669b603ba4","subject":"update constraint for python-magnumclient to new release 3.2.2","status":"MERGED","created":"2021-05-27 09:21:48.000000000","updated":"2021-05-27 18:44:28.000000000","submitted":"2021-05-27 18:44:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793328-new-release","meta_rev_id":"7cc4e95d352f094af721ad2d372e09cbb18f786e","_number":793328,"virtual_id_number":793328,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 18:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 15:55:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 15:55:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 09:48:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 10:18:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 11:12:18.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2021-05-27 15:55:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"7f7f79c7fc0c2998415c799aabff268767f6e5f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 09:21:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f65d8744053f8bcac1fe6cb5047736e92aa8f99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 09:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d0d7d3766dd84172a575b29264f8b2dd : SUCCESS in 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9da2490430177f4801a4b6da8595d59e47adce04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:18:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a8d85460e344b1bb1b314ee3d4d945c : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2145db3c49964b898fb16ae2637d46de : SUCCESS in 7m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaeccf3122624c79a4f92e0aca8e4b08 : SUCCESS in 7m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d67d81baf6c4419bbd4fbf29051a481 : SUCCESS in 5m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4823c1c39bf4f2c80dc1e425bffd05b : SUCCESS in 4m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dcf146cf3ec04a39a5b1032049d286d3 : SUCCESS in 25m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ecf1fb0d81a04a82ba4de2607895b3cd : SUCCESS in 10m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/495f8abac41d4229bb429115f4715841 : SUCCESS in 15m 09s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f42971520e164caeb6c3c402723a299f : SUCCESS in 6m 34s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/79e8a3c9faf34daba496c97691e83a29 : SUCCESS in 10m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/53682c2ecea44b43909aa3116e2f8aee : SUCCESS in 14m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ab884aee8c244aef8ce45c11ce27fa40 : SUCCESS in 12m 33s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/19fa1936d20c4bb0b7115309b5032aa8 : SUCCESS in 12m 44s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f9e0287de248493e97ba0b5e07bf1a8a : SUCCESS in 23m 30s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2ef943553b4e41189d4f27bb044da147 : SUCCESS in 12m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/497be2f55ba649d383af29fa95554e87 : SUCCESS in 6m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d1750635551241c2aec1721417d4543b : SUCCESS in 20m 47s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/13eb572aad4f40b681d4258076b09207 : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ad46c21b731541b7b6202e9bdd70338b : SUCCESS in 11m 35s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/83daef29c88d4129a6ea421d73dc5746 : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31c395d88c46468fac1b86e0fad21f6d : SUCCESS in 52m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/46a3b6076ac342caa4ac77a063f9818f : FAILURE in 13m 10s","accounts_in_message":[],"_revision_number":1},{"id":"e3d049ff890cca9a6f6d2439d91a2e301eb5a9e4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:12:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8c357fa28833aae1f03717b33d61e920c597533","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:35:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d7e95b703d0940339492948c2bea85cf : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":1},{"id":"a3d2c8e439d509133ed4362a1ec1bbeeaee3d069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:48:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c921d4fd6554b2bbcf5c0d63151e251 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbd27b39c97d45a7878e56eb7afc0fda : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf9a899a82d74ac8818dc0969519e20d : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d24283e25fdb4b828b01554bae835262 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/922cf24b92a74e8e8dad83dd5f373e64 : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/4be97c2b297d47e2abb7cb97ad160c35 : SUCCESS in 15m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bde057adebbc4eedaeff22f386011575 : SUCCESS in 8m 59s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/0443aabf188d4e84b49cf1aa073d5dd2 : SUCCESS in 11m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/fef05aeb39874559b561cb05f6e5e3c9 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/73ea8b9988e44a4fb659a1b3bfc17526 : SUCCESS in 14m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4a56f6854875484da5237eeca5304841 : SUCCESS in 8m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/320febe3b9d34e31a7537a72b9c115b1 : SUCCESS in 6m 41s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/306a01651ed6421c82ed69736fd06e8e : SUCCESS in 11m 57s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/42a09ce5d3224158b7dff03ec6c8960f : SUCCESS in 29m 08s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/fcb02b8658344d5dac2a5f81bed6442a : SUCCESS in 12m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3b8a980ee4954c44832edf013237391c : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/212c5be592424872b3f58b939a256cd4 : SUCCESS in 16m 42s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e58e42d093ae4e3ca58117475a826685 : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b195a8c4aee345c28f22c92bbc9eb58d : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/2f761e08b5d749c99e93c12786c700bf : SUCCESS in 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e150ce3e411c48ad836b79566e5107d4 : SUCCESS in 1h 00m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4de76b6611b4c99b9f9fa92cd80ed06 : SUCCESS in 1h 31m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a5fe0f14ed46ddee46fb07862d5f655f3dc1aa2e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:55:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4d7409f3fb3156f1599b57682bbc27addb298976","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:56:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae2bee370814638ebaf84e954973632eb7c76c1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:44:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2863c02eac9f406692d310eb9588bebc : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11acc65846844e2e8fe2b6b0839fa0ad : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ff1fd0c9c94494d89237a7a7e0df5f3 : SUCCESS in 9m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6978124f4324e17b4641a2dfe5fb9ca : SUCCESS in 9m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4058ac3b809344b4944573da84b6390a : SUCCESS in 7m 59s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ce6ee0914fb2447ba057374edb40c315 : SUCCESS in 12m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d479be39e0c14da59d7b732e4de09ad1 : SUCCESS in 14m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/caf17107322046ceb9718ec8570901a7 : SUCCESS in 12m 41s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4ef6ec861dde40a1a843570e5604daa7 : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c62c3031d8ae4df0a59d65563a49d174 : SUCCESS in 12m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb7086be47524c87b4cf17d36c3871d3 : SUCCESS in 11m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77efb26c1a1843dca83e70a06c19922b : SUCCESS in 6m 51s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/963b7d46922d4149ae141a47f1267050 : SUCCESS in 14m 45s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/f28cd4b43b554b0fb3bf7171ad1b77b0 : SUCCESS in 20m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/357f218ba55b4ff6bb450290c776ed35 : SUCCESS in 21m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/39e2a5b0653f456680f9bc6310db1dba : SUCCESS in 5m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/983c65491edb43dda8a4b49796c80c02 : SUCCESS in 19m 34s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/50e62220b1ea41adb770698dcefa2b7c : SUCCESS in 8m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/22bbab2d276345d4b192738c6e5da77d : SUCCESS in 11m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1653959536414a3faffe46ef3daa83bd : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0da5e07968374e3fa9fd63ce476e3f0e : SUCCESS in 55m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d87f6214276462eb1e3c0d8eca6ba59 : SUCCESS in 1h 29m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7cc4e95d352f094af721ad2d372e09cbb18f786e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:44:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87b2c6e81c58cd4a3bffdeee939461403fd78097","revisions":{"87b2c6e81c58cd4a3bffdeee939461403fd78097":{"kind":"REWORK","_number":1,"created":"2021-05-27 09:21:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/28/793328/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/28/793328/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/793328/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/793328/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/28/793328/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/28/793328/1"}}},"commit":{"parents":[{"commit":"378861b438e22ed6a5507516842689adb8e196f1","subject":"update constraint for neutron to new release 17.1.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/378861b438e22ed6a5507516842689adb8e196f1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 09:21:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 09:21:47.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 3.2.2","message":"update constraint for python-magnumclient to new release 3.2.2\n\nmeta: version: 3.2.2\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Commit: Tobias Urdin \u003ctobias.urdin@binero.se\u003e\nmeta: release:Change-Id: Iac2c12587db359530a7091ed00b6d738603a5af8\nmeta: release:Code-Review+1: Bharat Kunwar \u003cbharat@stackhpc.com\u003e\nmeta: release:Code-Review+1: Spyros Trigazis \u003cstrigazi@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I912b47ed6a91ad958dbe19242f3930669b603ba4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87b2c6e81c58cd4a3bffdeee939461403fd78097"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87b2c6e81c58cd4a3bffdeee939461403fd78097"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
