)]}'
{"id":"openstack%2Frequirements~793359","triplet_id":"openstack%2Frequirements~master~Ib1e1b135438eb5881980ad480d8c358ca6013942","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib1e1b135438eb5881980ad480d8c358ca6013942","subject":"update constraint for python-ironicclient to new release 4.7.0","status":"MERGED","created":"2021-05-27 10:05:47.000000000","updated":"2021-05-27 21:30:22.000000000","submitted":"2021-05-27 21:28:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793359-new-release","meta_rev_id":"d4a93d79b0f934bc864428fd2169edf0f30d77c0","_number":793359,"virtual_id_number":793359,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 21:28:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 15:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-27 11:13:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 15:49:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 10:23:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 11:13:22.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-27 11:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:49:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e8b2a4af3d8103b9782fdc493d9ed327f98f1b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 10:05:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"64848e7bd713b49ec98d17355e747d40e7f577e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:23:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/12d785ca93c64cfbb6de7e8f4365508e : FAILURE in 16m 10s","accounts_in_message":[],"_revision_number":1},{"id":"6eb4f44458c2d9a9c87ec95d4e9c6b585523035c","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:13:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"801fd0154a0b56001183374b2edb6f7757d610ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:30:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf6d7bf806d64591b3a6dd74835a0afe : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4d87f47ec364bc186a18818b787bb14 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46cbb81bb4884ce6a0174757e62bf001 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/993ec7af107245d093828d324828b6d3 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9258ef20436f485a8bdd1dcbdf55c495 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7aba0d45e1704d95be6186770867f1e7 : SUCCESS in 6m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e553321b3c645449998fa3d5deb7eaa : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f5e728f50176468db2a6cfeb66faa5cd : SUCCESS in 9m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4e061d5866f842d3b38d71f2ac3c7c4b : SUCCESS in 11m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0ec4dea65a704ee5be6fa24b7ae8580e : SUCCESS in 9m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4027f9410eb445df9a43ccaedda89ec0 : SUCCESS in 5m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0f537c964daa4a6c811cd5338fc3a64b : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60759d6874d649b187632027cc31ecb7 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a052a071e0e4c16930c51ca746c820a : SUCCESS in 14m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/81a0ff1bd4bd490f8e10a1b76e2083c1 : SUCCESS in 13m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/56f8f87380cd47e2be555d3d22bfab11 : SUCCESS in 11m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/87464d830b15407aa36b708a64c03048 : SUCCESS in 5m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/073e87155fab4545b2a9eb006876bc60 : SUCCESS in 23m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dee3422279eb428684d5e70d6e37eeb6 : SUCCESS in 12m 33s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b42565f1c4024135b2f2fed1debc7467 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/53253ca5b8af44ba9a63310bc5e63cc0 : SUCCESS in 21m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d8cc4a5a9c604cc5a9536276537ed942 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/86c278a946d14dbb83561b778a5fb556 : SUCCESS in 7m 01s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/45f2c19927ca4ad6bddc2c123aaf638c : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de228dfead7c49969572bef37688f24f : SUCCESS in 53m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42dba083638d4575b34c33ee5943cf7a : SUCCESS in 1h 22m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ec71da9906dca9680928e43c9b8f46af9d0d4f8b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c2fcb08efec4242cec648fe756611de75faea5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:54:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71e41c044ba84ec9b2c68b73711607a4fcb386b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:27:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd694026d507492eb1bc983d9d589f0c : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/614c7e6908e549dc92064f6495855d18 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dd63cac9c004d548901b0e2be5de4b9 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7713f8e6c794d5ea5a8703bca6881ed : SUCCESS in 6m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a544ae920844e75bdc5d74f632fdc26 : SUCCESS in 7m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/352768270a1d4d38832a215cf9723bf6 : SUCCESS in 8m 30s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/57b0c9e8b663472cae063028e75e7f80 : SUCCESS in 12m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a0328ceca1db465ab32ac92afac39780 : SUCCESS in 14m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/56b7b188c43b42c79de937b7f4dd7d16 : SUCCESS in 12m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/14c58167f4094d0e91af1913e1890fb1 : SUCCESS in 8m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/195d8b2f66fd4606a4b0b4fd26e18db1 : SUCCESS in 13m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e77533d3c3774e19a3d73af432b829bd : SUCCESS in 12m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c323fd2884714fd8a9e1897a496675c6 : SUCCESS in 12m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/925c9a0740c040608a4a9f6df899bcbe : SUCCESS in 14m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/66982c12c8a043edb43b792c2e96f02a : SUCCESS in 12m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/16d3116c54d44453b3b59d00965cb0c0 : SUCCESS in 7m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2c52896be633460ea17d2385ddcd966c : SUCCESS in 33m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/accf32ed8deb42b39608477b623903d7 : FAILURE in 14m 17s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1163244ad4db40cda57a21b8ec573a70 : SUCCESS in 6m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/78cf761ed9214fc998dd0e818a6207b8 : SUCCESS in 23m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d66ce8c09a81400b84ebc18964d3f5df : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/16e3801cb79449ad9ffa83087785e1ff : SUCCESS in 9m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cc52790fdacd4f69b7b508c9638d89ea : SUCCESS in 14m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3330621716a14562a80c94abbe800494 : SUCCESS in 59m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d180dc24dd5411cb1c37b35c4595412 : SUCCESS in 1h 34m 54s","accounts_in_message":[],"_revision_number":1},{"id":"877d4e00db6d38abf071f4e99f290cb2d30d872e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 18:48:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eae972fe925ba0cb2039770843f1b3cc46d0a470","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 19:08:08.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/22b448c686a849f3bf3609ddd8fade94 : FAILURE in 17m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f67d38057cdd9f2bf10e7895abdd4002f7828177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 20:13:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4607fe063cf744448018cd750a952baf : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc123b3c7c1142f88d077b176430768a : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/162c80c71aee455785505ea013d0eadb : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd57fa3df96f4400a52d4b8121d4a8c5 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f619c244419462aa241d527e507b62e : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/da247b42c8284af79078c0f4d2653b71 : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c34688eec4ed4e6c87cc8076ca3177e5 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/03080502268e41ac898a8729e14161aa : SUCCESS in 10m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/78a563e33bba4f508669680fa8a995b3 : SUCCESS in 10m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6d70340e1409469e9dd5f8790f042d55 : SUCCESS in 10m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d4d4475a94114c14b5e2792a47a705e2 : SUCCESS in 6m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4052b4aa106b4ce8862e6c1d0877b93c : SUCCESS in 7m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4afd5ed23874dd88b6dfaab786dd86c : SUCCESS in 8m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c31cc486024841be85f9f18902f26828 : SUCCESS in 10m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a92e26cd272748618fd200e935aba155 : SUCCESS in 16m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1ed51031b4b9487d83b6b683024fb598 : SUCCESS in 9m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/321b1356bbab4c778849f41f7083b63f : SUCCESS in 5m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8c3fefeafd174f3ab2552f0cd73ef344 : SUCCESS in 39m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8cc57e2d3ec44640bbfb40ee755fc86b : SUCCESS in 12m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4d44221f29d64530ba2828685e010b85 : SUCCESS in 4m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f4c327883b54d0b9ecaae621656845b : SUCCESS in 18m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1133e000b37448139c0d07f3840ff741 : SUCCESS in 5m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95a8cd8b359f44429c12ec03bc188a96 : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0c11240d236b46a3857c39c48fd98baf : SUCCESS in 11m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c55f3c8d1e4148d590411deab2b2f5fd : SUCCESS in 44m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fa932562452a40c2b3c1da5b309c52af : SUCCESS in 1h 24m 25s","accounts_in_message":[],"_revision_number":1},{"id":"37f30e83e6004cb0729f8e692a1be2a0c6a50a11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 20:14:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3d9ceb2311c8c8c2f66fa56b5f78d012c5648c88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 21:28:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e926f4975143c4961b4b51b16cd2f6 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/460bedab93534dadb78f76a845130124 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db40eb43063943f3b0bb032a2f6b1211 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe0e82956d2f423d9873ae1eeabc0275 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d6bb6da640b4168b0955a3489e64ebb : SUCCESS in 4m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5523b48c307c4d18b24fe7da9c19ffa1 : SUCCESS in 4m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6040bcaf95ac4136bd631a2359e8cc49 : SUCCESS in 12m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5a65e22604ae443493a0ba9b1a84d37b : SUCCESS in 15m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5390f4f1fcc3438683aec01a8805d370 : SUCCESS in 12m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/be0ad011dd7d4dd9aa79498e888288b3 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3292327125a74c11b1ab8e5a3aed58de : SUCCESS in 13m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3465b741bd6548b4a6b1a964253977f5 : SUCCESS in 9m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c1340d1cdc8e459ea43cd81491c40e4d : SUCCESS in 11m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6a0c745c51ad439b99d852395a2bae62 : SUCCESS in 15m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1330dbc41ca141f2882d22e03ba85197 : SUCCESS in 11m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/908eca75a8a94cecb3ae45671fc6e819 : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/db6ac3d18a26426197862027463ac445 : SUCCESS in 31m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dd6e9beed4844b14b0b21bbe2a10eb64 : SUCCESS in 16m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4df4e17543894617805724335b4b67a3 : SUCCESS in 3m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/269a2973efbf466fa040b05b300bf601 : SUCCESS in 21m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/81cdbf1203fd42a5baa9a7acb4ee5f53 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1131b8f47b064e05881e2e6c22e85ffa : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fd3de2902dc14f7dacc59208f67f150a : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddb73b4eecd7404088def0d48cc330a3 : SUCCESS in 55m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c16c1fa5658b463abcad3b86d129ed82 : SUCCESS in 1h 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f54c21545781b3da148b3d6c740e57754db3770d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 21:28:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d4a93d79b0f934bc864428fd2169edf0f30d77c0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 21:30:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e410091eb0d4603b9a297c6a26e2249 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5de488c08c4b06a17a7969b4dc9c2c574b0dffb6","revisions":{"5de488c08c4b06a17a7969b4dc9c2c574b0dffb6":{"kind":"REWORK","_number":1,"created":"2021-05-27 10:05:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/793359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/793359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/793359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/793359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/793359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/793359/1"}}},"commit":{"parents":[{"commit":"a1fa18d92350eddfc1621e11fb567e07e97b5b53","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1fa18d92350eddfc1621e11fb567e07e97b5b53"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:05:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:05:46.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 4.7.0","message":"update constraint for python-ironicclient to new release 4.7.0\n\nmeta: version: 4.7.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4179c48eb5ce19676c3e511410348d853d7705ee\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ib1e1b135438eb5881980ad480d8c358ca6013942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5de488c08c4b06a17a7969b4dc9c2c574b0dffb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5de488c08c4b06a17a7969b4dc9c2c574b0dffb6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
