)]}'
{"id":"openstack%2Frequirements~793363","triplet_id":"openstack%2Frequirements~master~I34db747481963a281980477a8be448fd594d798f","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I34db747481963a281980477a8be448fd594d798f","subject":"update constraint for python-designateclient to new release 4.3.0","status":"MERGED","created":"2021-05-27 10:20:37.000000000","updated":"2021-05-28 02:34:02.000000000","submitted":"2021-05-28 02:33:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793363-new-release","meta_rev_id":"a577070f30e64505e5fa169dfbf8a8079a1c512e","_number":793363,"virtual_id_number":793363,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 02:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 18:48:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-27 11:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 18:48:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 10:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 11:13:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-27 11:56:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:55:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-05-27 18:48:23.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-05-27 23:49:00.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"234c403e417b2a1ad728b3cceacbf212fab9c6b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 10:20:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05067861d63853c46051fa4664bd6aec14812afb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:38:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/764298b80d214619a039e5acd1b0abb0 : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"c14e4e540aa7c8ee2c0ad52c5a90cf3a5f1d99f2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:13:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fcd741d45f3fd9df1712bef7119e507f6318ce76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:56:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f6be32d088745cc9a9149eed83c7e50 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebecdde4016c4ce2ad8b16f02de537b4 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cced85481a141c6aeada1c3c89b47d0 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f95a598612e8426792a7b078aec1ae05 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/891095e7da9b4ceda84a8f5ab42ff4e5 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b59b34a648634aab8bc3b8059579fc04 : SUCCESS in 5m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c34096120a90492ab74e638d96b469fb : SUCCESS in 5m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/34e76820a6964655b927cf7431f95eaa : SUCCESS in 11m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5abbf5550875456c85738044c9913303 : SUCCESS in 11m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a24c2eff10994641966e326081460542 : SUCCESS in 11m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/49845b130efe4df581a2118d5745f669 : SUCCESS in 6m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d93fd0e1d7e84a36b3526f80066272cd : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66f41eae96134405a02f9ded6180a679 : SUCCESS in 9m 28s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/08f77b8d6fe446d4900ed027fdb71890 : SUCCESS in 14m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f7b4f21fd3764bc0a5b6f63bb32845e4 : SUCCESS in 17m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bfbedfdb50504820b0076bda5e736bce : SUCCESS in 10m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c7cfaaf0686c4065a23e0dbefc221f0d : SUCCESS in 6m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/37c1a0577dde4944947bc05827050d0b : FAILURE in 5m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3043d66f08d0441396b97a84a128ec6e : SUCCESS in 12m 51s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bbda47c967994cfe9af0d5ed0f49840d : SUCCESS in 6m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c4c2db8965a64578bd035d87f45d000c : SUCCESS in 17m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/54a09f10db4f4c02a86ee5911e08693f : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1d10fafb27a84dd191e5575b22879d01 : SUCCESS in 7m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/92ee1f25c1db4837952e34287a97e9ab : SUCCESS in 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/300abd5866664520854512214fd566a2 : SUCCESS in 49m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2e6d0f9de004426b9989018289b5894 : SUCCESS in 1h 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b0a2896c2517bd5b146ecee4595e988c1600b5bb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:55:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a93fa0773dc158218836ccd61f0e619c223712a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 16:15:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d9e8730cffc047358b6feecc41e371c1 : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":1},{"id":"25046afdd1e4f0f1d2c1fb4b2766c197b9954759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:02:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48fde25497894ffc9354629b2196b406 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562a6afdfe4944d1b1eea7cbe27ba86b : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cc3fe25513443f0a5d197ff6be46d24 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe307b40f0e4280a6cb0e24311510d2 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a20875df86d847709d60f3f58873ee03 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1e8c382d8954808a066aff995905abc : SUCCESS in 7m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9ec5f8cbeafc4049a20e037e7026f473 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9cbb4521e3d34b1e915a11f14004d49a : SUCCESS in 12m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/66338d4b076a4b72b798fa1cef370776 : SUCCESS in 10m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0ddf3edb7fa44285bf165f24bd82f3d8 : SUCCESS in 11m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1f03ac57cf3045848852a1217b298c03 : SUCCESS in 8m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f0e532fd08f947fbbd1547ecc7cea69a : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/297d1373f5544ecb9c41b5356befa6b7 : SUCCESS in 10m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8e7722273f9b4a01b81ad85ea01287f8 : SUCCESS in 13m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4d28448696af4b618ba9c21a0d2144ce : SUCCESS in 13m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/aff61358f5d5491ea4a2abcaf9674466 : SUCCESS in 8m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f74b36d291684b24a7455b1ccd58ed92 : SUCCESS in 6m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/22075b9afc2d4e638d3e0f0a0acddc0f : SUCCESS in 18m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7bcb1e9ff00946a393477ae35bcc4921 : SUCCESS in 14m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f6c2930db228401ba5ee9b2ee20813b2 : SUCCESS in 7m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/97bf59f0d87d495dbd3da3cc764dad6a : SUCCESS in 18m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2a8413c9ac7b4ac09a89d5c2de4d5087 : SUCCESS in 6m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6801f25916ea49cd82c06e95552280f0 : SUCCESS in 8m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4c912d654eba4a2d950740ac2096a3af : SUCCESS in 13m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b1bfd8e65194d05b33a30480366e445 : SUCCESS in 51m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dc9c6157b2254a8dae69d80d4241aed0 : SUCCESS in 1h 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"48e646a1cb258fdb14e6210cc07c76ac9e3894ca","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 18:48:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"171173d9d271771ccf71a8a009b1fa23fe30ccc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:48:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1eeb25afb84e49ca7d07f3e4ff617ab6494c9c4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 20:08:47.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb90eb62d9be4c2b890a427899fc1dea : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed8873cadac42b594a664319391404c : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43cd036940744d609be5d042bae3e710 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f7a00e8882847e5963c948887548686 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4397c35d640e4c288c43287744f019f0 : SUCCESS in 5m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6faebb42f42140a5b0fc14c034c836b0 : SUCCESS in 5m 20s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7a8ec3cc7ffd48928829c15d44939ee4 : SUCCESS in 18m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/032844eb86f74b8faef3cb50e1b368f4 : SUCCESS in 11m 17s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4b44fbbd2fdc4911bb6ecc94e4cdaef5 : SUCCESS in 13m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0d52c4b0ba8c42dfa5d34b77a1fa9498 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/46d832e74e4f4fd8be61f9fb1f3b5f96 : SUCCESS in 10m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/71d92fc0849f487481a830453b6f87ed : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/88fb64ec9f03437e99fc656a6d4f9374 : SUCCESS in 15m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/42f60d13a51f4ef79eebb80bfadaf09d : SUCCESS in 15m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/39549b104dcf48b9aab5a0cee7e5cc0b : SUCCESS in 9m 19s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/aa699f92fb6c42b4a8f29ca746e11e66 : SUCCESS in 8m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b4252ded31bf40c59990c5a704a0624c : SUCCESS in 30m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5bcbe572e22d46438cffb36175602f22 : SUCCESS in 12m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5712c7b012d84014b55a9090cd53d405 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2e679d63c5d54ebeb4783e8402441dc9 : SUCCESS in 23m 57s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fd44d57dc6e54c0895eb71c83ad7ae72 : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8fefc7ed6384759a428f73df8547683 : SUCCESS in 8m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8a3318b820f743299d0e512fab67379a : SUCCESS in 14m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/686f78fea6dd41c5a50782a073debe95 : FAILURE in 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f4132e1157344f4ba27876a7c4161f6 : SUCCESS in 1h 17m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a0d58558950a99900ddf5cab8b6836cb8d0c7843","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 20:50:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7630639b9aef7739cfd6dc8b67d04f4304f5e94f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 21:09:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/11f85e98fb634be8922a5422c29b5a83 : FAILURE in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"07f667e2099aa1095aaf1542c01ec5e2dfb79991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:37:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f8cf7212b8e485d868fe38f33aec3d1 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ccd727eb9c4626ac834b871edfc537 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/adcce61e490b46adadcfcb7c77f1a799 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f46cd45c06b48468b79bd9d64a8871c : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0089e9dc1f84120b1511f786263aec1 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ac35be4843ef4fb2ae10afa5f7b2fb09 : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b073aefb0d354a62baa4fcb4dae72231 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c0eb90c7365a441cbcdd69774322c71f : SUCCESS in 11m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c50a4860b23843c3acd07befaf3d34d5 : SUCCESS in 10m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7da2faf1c8354a39afd3de104c27b4de : SUCCESS in 11m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9914035b5ca546cd8647cdcd4b26e3ab : SUCCESS in 6m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/78cdf8617bf048a8b08847254acbfa6b : SUCCESS in 11m 46s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2a59c37603c54f1983864503a4e7980f : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/76fd9ac3c9cb40e3b81231b2b0579ea1 : SUCCESS in 11m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/897489304fca4351b7113de262efde41 : SUCCESS in 14m 02s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/05a9ffbbfa2e482fb2b14fc64f7ff326 : SUCCESS in 8m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5308a8e284bf41348d20207ae0718b5f : SUCCESS in 5m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8deaf51e6cfa476b9a292df04cd2d680 : SUCCESS in 29m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/954481d7008e4fd2ae4c665717c6e981 : SUCCESS in 15m 10s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc9ab3034e844ce89a45cb6bbba28a4e : SUCCESS in 4m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dec647c5d20d436ba8558a7ac03bc061 : SUCCESS in 18m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/75a7be1d231f4ed5a4d22c5833451a97 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/59cc0103d87b403b9bc5d051cdf7a49f : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/844fd5be6400449ebd88264ba073609d : SUCCESS in 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1397b49f96a40f4ad12dfb6563042cd : SUCCESS in 52m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f85a1d9fe06046d683db9de9306f3807 : SUCCESS in 1h 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e178902ef37a73bf07ebd6208237ce672b0daabf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 22:38:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"13003a496e5fec0a0bb960621890779486e120dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 23:48:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f1c18743724979befde7b73a60fb61 : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c4b1559d8cc41b8a1e7aed9c408cdda : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dcdb5a54fa44f7fb7c2d537019fced9 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b0224723075493cba38fa1605356f59 : SUCCESS in 5m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d37256021e749ccbff3c65c348e78f5 : SUCCESS in 6m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc19f34bde3b4f778e4605be53016131 : SUCCESS in 5m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/65f2e4c63b6f4b5589d760d89597d5f1 : SUCCESS in 10m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c74908c788134fcaaeff338b6ae0d5b4 : SUCCESS in 11m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b9145c228b16455ca1e2f807ec692e38 : SUCCESS in 22m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f4a273283f664b0e8597a001958a9ad3 : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/559fac866ba24272b5f5ccf8e5ef0cc5 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a898590ae77480b9e557883209f0e9d : SUCCESS in 9m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/224d1f1cc9c24bbb99d2c5f8ff2a31a2 : SUCCESS in 13m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8011055f4fca4dc1a70faa944f35af13 : SUCCESS in 15m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4229338b15a34cb5928dbe0131f55166 : SUCCESS in 11m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b300ba9407fa4ed5afd1ab8ab0a537e7 : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b2de2cc56ad74d0d9d88b4a961b18370 : SUCCESS in 24m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/48ae8f33291542ca98c8bcd335f0f4b8 : SUCCESS in 12m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/adefb2e10e0a47adadb92757c620f0e8 : SUCCESS in 4m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8f58d37c5eac468f82a006c260bb9d07 : SUCCESS in 14m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d3d6db6c838345cc99a9671ac0f47c7c : SUCCESS in 7m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2320d24889e40c5ac641c56cff1bb13 : SUCCESS in 6m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a7daad489ea240c3b6195b9bc3a7876d : SUCCESS in 19m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9752eda2699645a8ab2a7276b239ad5b : SUCCESS in 40m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e6b26e63686d42e3b3cc6d315940c959 : FAILURE in 1h 08m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e43c387b63b8e944116a12f2e6221ceed232b552","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2021-05-27 23:49:00.000000000","message":"Patch Set 1:\n\nrecheck\n\nHappened to be watching and saw the tempest.api.object_storage.test_object_expiry.ObjectExpiryTest.test_get_object_at_expiry_time failure -- we might want to check the assumptions on https://github.com/openstack/tempest/commit/53ce5ae4d662ca2ba961c6d9e3af0a3a89a1be5c; I suspect some calls can take more like 8s in the gate.","accounts_in_message":[],"_revision_number":1},{"id":"efd263e4b210c12d7811ab674238d54aebe52801","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 00:11:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/293f067b9d4541019d51c91ad20c704d : FAILURE in 21m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1e5f059f037d2a8a824b12d520eb367bc1c97245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 01:18:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/155010ab2f2f4cab96b553c346f46e50 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f9d42a5cd5646488242fe2c73801b21 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8d278b5e86b49ad904864ccc38d15af : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47389f0f6d90407090ec31da84b2bd88 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38931b268cf243679a65c7a852cef68a : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/385cf3f123a54638b4bcefb7261acf69 : SUCCESS in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3f71dc8f6ee243c091ed37ce755da7c5 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6cadcbfb59e44fbbbae61bbb12a4dee9 : SUCCESS in 8m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1f5653870e5243788a3909545d69b929 : SUCCESS in 11m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7c7fe16d42e94c178d907204c8662ba0 : SUCCESS in 9m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f03b77050b6740179ce4ae792db7fc5b : SUCCESS in 6m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/39be204941ad44ebac4bcc2c2dc80db6 : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d90fc228deb1424dbf3286c17906fca1 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a789a46956254d4db590ce91fa97c838 : SUCCESS in 9m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eca5c4ba0bcf4cfbb7e248941a9dba31 : SUCCESS in 21m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9106239703734d3fbefc967989b3747d : SUCCESS in 7m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7f11b103fc004a95b70cf8cbdd934e60 : SUCCESS in 6m 11s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e1e9531161bd42509ab243d032d062ba : SUCCESS in 27m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2d41df327c864aa8bbff876f5c0d612a : SUCCESS in 11m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/044167c4d40a416da5e8e06742532a5a : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/40f68d9aed474ece8f93900b7701b1cf : SUCCESS in 16m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d7dee5c26ffe4b658bf7c5e80b359e4e : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/72299b521e41412d8090e0fc5abcaa65 : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8f0ae7499534455fb2cf39096a3d127d : SUCCESS in 13m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08234c4eb2dc4df2ad6720bfc3498fe2 : SUCCESS in 55m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/57ce8f722f1b4d37920f53080e1bc2e6 : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":1},{"id":"742998662e821f5e16559093a046bda6bc92a895","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 01:18:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"edcba99245f6bdfe5b114e1c700cff6e184d6acf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 02:32:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/295bc0b316f94515ac2cae8b83058569 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f76addfac34f80a6262c691a933f61 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00b9240d3477450181ea386fd1634023 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/faa72c809f9840b0a5e9c7dc447bb7cb : SUCCESS in 5m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d4d54c9b494435293f345c59e04a40e : SUCCESS in 4m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6afa0aa5b8454cd0b164245c99ce9e45 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1feb4c8d2be749599dacffd3f5135b99 : SUCCESS in 18m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db25f381f3f046c592aba6605455dc29 : SUCCESS in 12m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/384d8b2784604298aa8643cecf9318e6 : SUCCESS in 12m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a3250e6f61c14cabb04febce4cc0d1af : SUCCESS in 7m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f81c93026b0f4709992db70ed5236a98 : SUCCESS in 11m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1e9d4ce2c254fad8c8cf204cd2afecb : SUCCESS in 10m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14ff200568cb448ca693579938eb7235 : SUCCESS in 13m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4497f7917ac542059af79214a04d79da : SUCCESS in 13m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4c2fede294ac4d2eaa5dfd9c2713284b : SUCCESS in 16m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/11a2191514be4e9183e4fa02bdd3e752 : SUCCESS in 6m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ab9cc0f09a434030926f99b5f7193972 : SUCCESS in 23m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/81310e88aa7a4c4bb1a8f3bd3654a7cf : SUCCESS in 15m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd450926a9c1417588da7fecbe594f01 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b331054329034ad9a8b1ca9ab674af8c : SUCCESS in 18m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0116424086b7450facbc06c815606b23 : SUCCESS in 6m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/895bfe9cdb244c6d8e729048566ca7e4 : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3ddfd9013cf4a96bc33fffbf106dec8 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a68f1316898f4bbd9acb9fd1780a8f1e : SUCCESS in 38m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7de71946d8f54755a7abc3d191b18156 : SUCCESS in 1h 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"890c6b75200e5c1723ceb393949d684b954b09b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 02:33:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a577070f30e64505e5fa169dfbf8a8079a1c512e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 02:34:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40bd40b30bf7482eaa8b7ee590a0ce71 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7a19b3f307c97c34815545ff668308d879c2253","revisions":{"d7a19b3f307c97c34815545ff668308d879c2253":{"kind":"REWORK","_number":1,"created":"2021-05-27 10:20:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/63/793363/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/63/793363/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/793363/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/793363/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/63/793363/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/63/793363/1"}}},"commit":{"parents":[{"commit":"a1fa18d92350eddfc1621e11fb567e07e97b5b53","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1fa18d92350eddfc1621e11fb567e07e97b5b53"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:20:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:20:28.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 4.3.0","message":"update constraint for python-designateclient to new release 4.3.0\n\nmeta: version: 4.3.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I845d699e26cfff3d63e8c588a64202c92736824f\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I34db747481963a281980477a8be448fd594d798f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7a19b3f307c97c34815545ff668308d879c2253"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d7a19b3f307c97c34815545ff668308d879c2253"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
