)]}'
{"id":"openstack%2Frequirements~793364","triplet_id":"openstack%2Frequirements~master~I7dd42784726b0a9c07cfd68337ec4524343ed5dc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7dd42784726b0a9c07cfd68337ec4524343ed5dc","subject":"update constraint for python-neutronclient to new release 7.4.0","status":"MERGED","created":"2021-05-27 10:21:00.000000000","updated":"2021-05-27 18:32:02.000000000","submitted":"2021-05-27 18:30:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793364-new-release","meta_rev_id":"6b94c308b0120175c69121b60c5760fcbbd19c8b","_number":793364,"virtual_id_number":793364,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 18:30:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 15:50:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-27 11:14:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 15:50:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 10:43:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 11:14:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-27 12:02:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:50:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ae5e8b3513779459271392349aa9bb26647e5e5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 10:21:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3603e7cf0df83a09df002250c9f4b5c72db9d0a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 10:43:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8438fcd8a0c6414bbf4251843517d94e : FAILURE in 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d619759838bbdb5adea6ed566ff0370661bb853f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:14:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"eee2af2d6c3af1288125d68889a38a1341c79a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:02:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72cf7f094fb14f938f73b1c5362fd7dc : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75b9c775895040f4b9eadcb062e45f7c : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e70ad320734a718baa84294f7e2212 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef24855a275c4e02aa9b12c132ac30bc : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/259dfa6d8221455a8fdc9215469d9880 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1007fe75cd9437e94a7c8cbe305f02b : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5fd929084d5c4597bf317ebd6376bf12 : SUCCESS in 5m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/60000083b3b24534bcddc8fe801ed156 : SUCCESS in 11m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d9cecfa77fb48e08043864868708a28 : SUCCESS in 10m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f84cfbffd1dc4df381c7638295b514c1 : SUCCESS in 11m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5de809abba654999b4c1e93a968ef0a8 : SUCCESS in 5m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7dfd8585d3e44051ae0632966cb39523 : SUCCESS in 12m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/991fd70ff76f47feb258d808917be4fe : SUCCESS in 9m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/27d90ca733bb41af98cd325d0a55d379 : SUCCESS in 16m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1712f23d82cf4d5083a14a3b19921fe8 : SUCCESS in 15m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/01cdd44e609d4c0cab73235c599ee6aa : SUCCESS in 10m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/62a7bfecaaf94f9a9e881b9d32a8871c : SUCCESS in 7m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/615a7677c78c4d4c94d2a2ca9f1b7c9e : SUCCESS in 29m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ca6c3862146049d18279514798880cea : SUCCESS in 13m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bec6ecef31854deb89c5234dc7b5b362 : SUCCESS in 3m 36s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cbf94be210d44d0785a16fd3d4599381 : SUCCESS in 17m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/04976fae36b949b2b9d0046716616251 : SUCCESS in 7m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8688733842c34550b639eb842f6a752f : SUCCESS in 9m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3bfa0b5abdce47a897e5a10ca3c2b919 : SUCCESS in 14m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/654ebf6d75874f879f28cee3d17eafb5 : SUCCESS in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f27421d5754944f1880de076e43ebdb2 : SUCCESS in 1h 36m 14s","accounts_in_message":[],"_revision_number":1},{"id":"02414c69ea5ece823f57ac902201d0c6d9ac3596","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:50:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"263768ce4bcbbae7d9d43ed27bc9ec9c3a3164b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:55:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"937d5e3956098328479369e990325905b6ff9b05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:30:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d27256f9d94c369d958232fa7452d7 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5533503e28cc4f02aa633343700170ec : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dec9f1f59b214011a4862ff5cb89232e : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07882da4619e457a99781157d03a66fe : SUCCESS in 6m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ca8ef19eb8a45f799c561eb2508aaf7 : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8d19d10eecc349a0b8562d50d0d8baca : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1a99918278e54a21b9916d299a37e5cf : SUCCESS in 12m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b1dee2744bca476aa2e62c330cec1e72 : SUCCESS in 11m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2368ec9af0f248ba95a31ae8afaf578b : SUCCESS in 12m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/335f2ad339e047be8deefa134c055642 : SUCCESS in 7m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fbbacc0022c64cd1acf905e8bc87eb95 : SUCCESS in 13m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4b3eec1ca974813a87be78622db2324 : SUCCESS in 10m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/99233a089e8442bfb7a4a1fd127c8fc7 : SUCCESS in 14m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/069bc4e2cee94655a0f3e7d0ce637a9e : SUCCESS in 17m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c8d2e60e5389437282fc04d5de04ddd4 : SUCCESS in 12m 29s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bd1209af70d64aeca5166e6cf765f79d : SUCCESS in 7m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6348a0f6b8cf4104957ae2dc9fce6be0 : SUCCESS in 20m 35s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/149392cb7dfe498d822ac14e189181cc : SUCCESS in 14m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/89b704ff950e49118e306812198dac64 : SUCCESS in 6m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/111f75c6b0d9476a981afb649a936eaa : SUCCESS in 18m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2bd1de9d18464406b9b166f0cf4561e9 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c1726d1d9b9948a2b95d1a8f5d7793b1 : SUCCESS in 9m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e30f5f6ce40041a3b5187a99b26b6a5f : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81aae9abcc3f4ff6b7839f23e2be1cf3 : SUCCESS in 56m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05dc5ae7309a45e897fc2fbd484ab7c1 : SUCCESS in 1h 27m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1fe8615eceaf29f1da7c83583346bc02a0e90a74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:30:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b94c308b0120175c69121b60c5760fcbbd19c8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:32:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bff1c75aa094451a614b1bce8a1c5df : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10cde85a2bf5285c451199739eb6e9061b2fa06b","revisions":{"10cde85a2bf5285c451199739eb6e9061b2fa06b":{"kind":"REWORK","_number":1,"created":"2021-05-27 10:21:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/793364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/793364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/793364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/793364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/793364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/793364/1"}}},"commit":{"parents":[{"commit":"a1fa18d92350eddfc1621e11fb567e07e97b5b53","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1fa18d92350eddfc1621e11fb567e07e97b5b53"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:20:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:20:58.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 7.4.0","message":"update constraint for python-neutronclient to new release 7.4.0\n\nmeta: version: 7.4.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib81ce3bae5c68d1d61b2a16b43f746e3d8e08b5a\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I7dd42784726b0a9c07cfd68337ec4524343ed5dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10cde85a2bf5285c451199739eb6e9061b2fa06b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/10cde85a2bf5285c451199739eb6e9061b2fa06b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
