)]}'
{"id":"openstack%2Frequirements~793375","triplet_id":"openstack%2Frequirements~master~Ie52ad95167ac6f6496ee2c1194569abc9910b09b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ie52ad95167ac6f6496ee2c1194569abc9910b09b","subject":"update constraint for oslo.upgradecheck to new release 1.4.0","status":"MERGED","created":"2021-05-27 10:50:06.000000000","updated":"2021-05-27 18:43:34.000000000","submitted":"2021-05-27 18:43:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793375-new-release","meta_rev_id":"8609e9d1ab60350d73af8e8353d1a696b95ca8cf","_number":793375,"virtual_id_number":793375,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 18:43:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 15:50:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-27 11:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 15:50:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 11:12:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 11:25:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-27 12:27:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:50:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f7d45a1a56bd1c91d303cefa9175adda7972df0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 10:50:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"844afe7407f1c8ed9eda4a1f8b4b66c4bed98b25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:12:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c4c5bcce1b8144bbaaf6799fe6d62006 : FAILURE in 16m 35s","accounts_in_message":[],"_revision_number":1},{"id":"17209a0f59cebff36c155858afce6a305f44af2d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:25:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4b1eb1e4326c65e6cfc0d56a38e4ee51a1d51839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3d4b72b09c4563924f2b45164b3741 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e178d6b4374ef9954eb923e53b172c : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63cc2b9fc4cb48f78952697af87a495c : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5657ad81fbe451aa9071065cad61c51 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bb180290e7f45ee9b8cb792cc1d1825 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cba1dde8462d4e39b24599b29da77424 : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/643220921d054661af02c95a9bc6ca4c : SUCCESS in 4m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e332c10fd8b24aee8205377aff2e0c47 : SUCCESS in 10m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/48dbfc5269634e1287e91299e6aa5201 : SUCCESS in 12m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/501978ebe96f4598be2bff2156ac27bc : SUCCESS in 11m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8d4d3e51fb8643c681ca126124d0f540 : SUCCESS in 6m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1032f9ec634242ad859297de1f45e4f6 : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f3b30374755e4eaeb0e24d5b9e5d0083 : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/97d620e8ecb04072961479bcdd17f951 : SUCCESS in 14m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7ccd7c6297cb41cd8e725168cfc5762b : SUCCESS in 14m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8e371614a9ac45aa9a985397478893c2 : SUCCESS in 8m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/33173e06296e4fcb8a00d24844d31899 : SUCCESS in 5m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8c31d604dbe74d5e974bec2202429790 : SUCCESS in 29m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a3163517224c4538b967e4ee3a3f7c3d : SUCCESS in 14m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/dab9dd1a826544f68644c34eda2ccaed : SUCCESS in 5m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ce38796c5dd84dc9b478f454b89ad56b : SUCCESS in 20m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/16e05c02769e4bffb506d80981d79f55 : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c5e295938f0453eab866df0fe518c94 : SUCCESS in 7m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d2937122832447e391dcc5d16678e0de : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dcea9dad81524d74b9be901110ffff81 : SUCCESS in 55m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/721a2bff85bd4915a005c0bff7fe666a : SUCCESS in 1h 33m 37s","accounts_in_message":[],"_revision_number":1},{"id":"958a579b74a26a5e58d646cf7d8240bafe99737c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:50:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d056605431e466deb5d5aaa335fae0f165ba623c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:54:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28c7884a667c54122686603478600b4910b15d99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:43:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dd8c1e52c534dbcaf7421b105ebaf91 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21dc110f2b3e4c1c9d33c3000ee5d515 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/792d0dfca73c442aa4940eaa0349e086 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77dc50f42b3b45e69a82149429583860 : SUCCESS in 7m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/abcdaff364c347dcab9c38de3174ceaf : SUCCESS in 6m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77be6efef46b4cebac17ed357914a073 : SUCCESS in 5m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/59b87ae05507446f992be8f3c35a91fe : SUCCESS in 11m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/68a3407eb324455d8e0fb276f144b44c : SUCCESS in 12m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4ce4914675b24f069f513370b75157b1 : SUCCESS in 11m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86d6fa926cbb4fd3940797449e5e544a : SUCCESS in 7m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a9f2e36656054df3a00b72700caea56d : SUCCESS in 12m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49eb56771fe64fa6a665c68238b7190d : SUCCESS in 11m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f5001d47aa914ba1a335e9cc7ce7a6ba : SUCCESS in 13m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b605d468706741e5bcbadeafbc48fb10 : SUCCESS in 16m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4855bb0c982b4117a1f415c612c420d3 : SUCCESS in 10m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8de521b150c2469d933de8e85dc87ecf : SUCCESS in 7m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f598ef4a625941838d4b58518ab9cfdf : SUCCESS in 28m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5527cfca5a6644ef89933d693c7d15ef : SUCCESS in 14m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e07bc9ae44e24119952fcf49b7268c6d : SUCCESS in 6m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/345f4e7eb3ed400f962d03eee49e6814 : SUCCESS in 21m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c86b23afe8aa4529bb3417b4694adaa6 : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/949fdcc532d3444db9d621e5652cbbe7 : SUCCESS in 8m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5b2980b8f15d40179c3e6dd4f1f7bfdc : SUCCESS in 13m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0679f387620406ea447dacfe695d3b8 : SUCCESS in 53m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce365dca02a24cdbbcc74a1bd07dd721 : SUCCESS in 1h 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8609e9d1ab60350d73af8e8353d1a696b95ca8cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:43:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51e4514860fd144918d746f22a6bf0c62d756e23","revisions":{"51e4514860fd144918d746f22a6bf0c62d756e23":{"kind":"REWORK","_number":1,"created":"2021-05-27 10:50:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/793375/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/793375/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/793375/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/793375/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/793375/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/793375/1"}}},"commit":{"parents":[{"commit":"a1fa18d92350eddfc1621e11fb567e07e97b5b53","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1fa18d92350eddfc1621e11fb567e07e97b5b53"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:50:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:50:04.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 1.4.0","message":"update constraint for oslo.upgradecheck to new release 1.4.0\n\nmeta: version: 1.4.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I08b56d3dac10382af01b02dc8240ac4ad9734e6b\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ie52ad95167ac6f6496ee2c1194569abc9910b09b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51e4514860fd144918d746f22a6bf0c62d756e23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/51e4514860fd144918d746f22a6bf0c62d756e23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
