)]}'
{"id":"openstack%2Frequirements~793376","triplet_id":"openstack%2Frequirements~master~I26def4d8bca9f211dafdd84e7394ddc8e4ba09fe","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I26def4d8bca9f211dafdd84e7394ddc8e4ba09fe","subject":"update constraint for cliff to new release 3.8.0","status":"MERGED","created":"2021-05-27 10:50:42.000000000","updated":"2021-05-27 18:43:39.000000000","submitted":"2021-05-27 18:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793376-new-release","meta_rev_id":"e509edc202303b1734e2233fbcd391140cfc4bdf","_number":793376,"virtual_id_number":793376,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 18:43:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 15:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-27 11:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 15:50:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 11:12:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 11:25:30.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-27 12:29:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:50:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"53a4c7c6db4ab9f141ab378fa36a77a8925a77b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 10:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"499ca7dbccdc07c24f01e40ca8620ea2923a09e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:12:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3ab1dc8edc294d09b0a260b8be8ba9e5 : FAILURE in 15m 37s","accounts_in_message":[],"_revision_number":1},{"id":"d9d9df5303a63b96f288c7f1ec66eb7d1d261fb7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:25:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bdea90d6438cb6f4323a73a9a391681c185d599f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:29:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b592a8050fcd4a7c8361c7fc1e0f8c9d : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bbc5aca488c454e876744eb4000ee48 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2201c002cff438e960bf9607abf2864 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95c8a395419d424fae5335b115ea900a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8704513079343459324129094d43c96 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/136a8062b5684c2ebecc939e840ae7c2 : SUCCESS in 4m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/170298ae60e447a7b87ec56df3deb8bb : SUCCESS in 4m 33s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6661d5c8e5dd4673996cf9148c1568e7 : SUCCESS in 13m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/846952ffbc144fc191f424e0005da39f : SUCCESS in 9m 51s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4f6a424bf72b4f94b5c3c0d23e9ff802 : SUCCESS in 11m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/51317743e3244f3ead3b0e90066f1b8c : SUCCESS in 5m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/936148ea6e68416f840b023524d817e9 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6548ee844240452999ec52e56d734d65 : SUCCESS in 7m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/582297eafc0944f89bb9a2082fc3ff23 : SUCCESS in 12m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5508963188a848a5aca8e653c71343f3 : SUCCESS in 14m 23s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e583903855d34cfea3cd30cf4eb52769 : SUCCESS in 11m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f3d5e80dbb4745bc8efe41bd1669e933 : SUCCESS in 5m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d7fa86c18a5f4073a22acb3a6f3bb163 : SUCCESS in 28m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1c0d51a2f05c48a9847144a80bd87c36 : SUCCESS in 14m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0b0d63d93f084d7aa320ba5492f4500d : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/453491ad34c046aba4c6fd5afb8175ae : SUCCESS in 19m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/318887e9e3a2461992baa2b16a93752c : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b92c5e3eaa674a5eba70f0f3b19b4d38 : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/017783269fb34c229357ae551e230a07 : SUCCESS in 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d6bd04f5da64ad9964123cdd18a840d : SUCCESS in 53m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60d806b3ceed49d89a7b918e1769130b : SUCCESS in 1h 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"a0cfebb0af4dfeba54535f3a9e96bd846ff912c7","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:50:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"90dfb1a8f08e1b5e992dfc74a624f6eb3b8cc0e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:55:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3a11c84ca3b1c9f4b615a2af9fba716905d7bd98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:43:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f083dea35ec944ba8c389db9c9c78979 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d34a92ef4e49f9affb4095a06d0ca6 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/748594f842bb4aca9c5d52ef64bb181a : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fda3410f9c5c4c1aa6f9feadfa411e61 : SUCCESS in 7m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbecb0b027dc42ff9d2147b10cac7655 : SUCCESS in 6m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f7c366ca67a446be9ac6d7d03f2468fc : SUCCESS in 6m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b9e78bd7366f493180c1f4ebf462e2b6 : SUCCESS in 12m 03s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c8f225b9bcb4ca587d24d1dd622fb8e : SUCCESS in 13m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/38ee3a5dcb554db7b1f58bff4655f3f9 : SUCCESS in 12m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/54e13508eb6d4a73a8cba2c355a9c86b : SUCCESS in 7m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/83bc5b6a95944c2e8be2a2471c99fdf7 : SUCCESS in 13m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/62dabe591e2a4a62830e44dc364a5ebe : SUCCESS in 11m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/66f33aa346ef48558de9f27f033e6919 : SUCCESS in 12m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2548d4bb7c2a4583b9c88f264618722d : SUCCESS in 17m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ab68e64d376440b69515ef508471e0be : SUCCESS in 13m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/012469813fb546109ef0733b4f6f5f3e : SUCCESS in 7m 28s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb8ae558886944ea9a1668c3fd70825e : SUCCESS in 31m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f2aa1804bb064a28b66b4aae08bde1fe : SUCCESS in 15m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/98bef93048494c35ad0de0186bd82e2c : SUCCESS in 6m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa57daa30eb0492e838a6c6bf35b551c : SUCCESS in 23m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/30289898ab3f4554ae30e3cbdd7837c7 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee2f1b24bcec4716a5f7699ae41d8c41 : SUCCESS in 8m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ec836b1ab4234162898d7d6202d3f2cb : SUCCESS in 14m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/576b4fd140264ab7a8f1df0149db670f : SUCCESS in 50m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aaf47d43472b44658949bbfd334ad627 : SUCCESS in 1h 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e509edc202303b1734e2233fbcd391140cfc4bdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a032e90e4e071caeb1fc6f5acdd7e43eb641fec9","revisions":{"a032e90e4e071caeb1fc6f5acdd7e43eb641fec9":{"kind":"REWORK","_number":1,"created":"2021-05-27 10:50:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/76/793376/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/76/793376/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/793376/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/793376/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/76/793376/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/76/793376/1"}}},"commit":{"parents":[{"commit":"a1fa18d92350eddfc1621e11fb567e07e97b5b53","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1fa18d92350eddfc1621e11fb567e07e97b5b53"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:50:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 10:50:40.000000000","tz":0},"subject":"update constraint for cliff to new release 3.8.0","message":"update constraint for cliff to new release 3.8.0\n\nmeta: version: 3.8.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I8d3c3c932dc5c91f5e0476e14030124307991ae2\nmeta: release:Code-Review+1: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I26def4d8bca9f211dafdd84e7394ddc8e4ba09fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a032e90e4e071caeb1fc6f5acdd7e43eb641fec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a032e90e4e071caeb1fc6f5acdd7e43eb641fec9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
