)]}'
{"id":"openstack%2Frequirements~793381","triplet_id":"openstack%2Frequirements~master~I3c562e90f1b690e1f6c559c01ac8e685c7623036","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I3c562e90f1b690e1f6c559c01ac8e685c7623036","subject":"update constraint for oslo.cache to new release 2.8.0","status":"MERGED","created":"2021-05-27 11:13:26.000000000","updated":"2021-05-27 18:46:37.000000000","submitted":"2021-05-27 18:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793381-new-release","meta_rev_id":"1f7436bb55333708b1039865c420935b766effd9","_number":793381,"virtual_id_number":793381,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-27 18:44:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-27 15:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-27 11:26:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-27 15:51:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-27 11:26:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-27 11:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-27 12:38:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-27 15:51:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5954ee0f0f58ec904cfc3a3b9b62796e9c7262f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-27 11:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41766c453f98cbedb0d805c955cea477e60c2d68","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-27 11:26:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"038a17546691108f5945357162cf7882ed0b5353","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 11:34:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/71d5bf569add40d983ef557507fa0f8f : FAILURE in 15m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f658b3515fcd826643dfa54cd9758be3f1ecc76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 12:38:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/647fb62f125a422d862ef3b479c40727 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9518003f7084e7b8023156687b03372 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/585786015f4e4fe09fe48cf8e14079be : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a75d603e8cb44dceb9c670c4b85d9422 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8237df8a724e47db8187aa0927c00526 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ffc4b329e244241b4364ec8ea15d221 : SUCCESS in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b91e566af8fd403997009342bea7c515 : SUCCESS in 5m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a0612c21233c4b7098703535e763e35e : SUCCESS in 21m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5fb51b32fb6146c08de7c1f857d574f7 : SUCCESS in 10m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/683ee82020964af69e3dca53917c9d01 : SUCCESS in 10m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d72a500b9e3143a3b7983f6c4d13c09f : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3983cc7e74cd45968b0839e46c488e19 : SUCCESS in 9m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/256f34aa69b2445cb371a398e09ec158 : SUCCESS in 14m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e34bef089a83404c9f1af3f9a5b4cc36 : SUCCESS in 15m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d1814d8fa9a34028b6ef1d46624f543b : SUCCESS in 11m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b7b2141f33c240418d53a8b9451642c4 : SUCCESS in 11m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/27658e94a008455ca1f34e47e93d300c : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/49aae2c96d664c68a21ff9ca78c333e4 : SUCCESS in 32m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/faa0d3a9ef0940a0898a816c62bfcfeb : SUCCESS in 16m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/743783649584460482af071c4e18ac7d : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6b98ce7264c54be1bca2d21edc5ad18f : SUCCESS in 17m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8e5cf99701974224bb2d7e39826b15d2 : SUCCESS in 6m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5fb54ecedd6f4e27b76e8b554d4d975a : SUCCESS in 6m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6d776dfbc71147abadb445a2dafceb38 : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e155b846f1b44343b0e8688b546a3723 : SUCCESS in 52m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b7007451add64596a70aa6f88bfc3085 : SUCCESS in 1h 23m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a039bfa7b6d6896ff78b875edcdcff20503d854c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-27 15:51:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f021e5a4194833426d974a58b236feeb190b9a94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 15:56:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"155b3d0a6e154bed7eaedf13824f1d650e052006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:44:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d3fb1c757e4fc2bf18199c434a5e04 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f784588ddad94f3c8345e9bad065fdcc : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fbebe418986145e5bdc3470b94c6ad94 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a846e25ddcc44f4c8fa3b2c16d54db65 : SUCCESS in 6m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/347792ae955340d9b7f1c474512b5a0b : SUCCESS in 10m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3228022ed4d14c79871e5a59f8655ad3 : SUCCESS in 5m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93280f9e67ca49d9bfa7ea34c4646dce : SUCCESS in 11m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/281bbc4bd9914e9eb7ce4f6f560cdc6c : SUCCESS in 10m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f551eb3403704c96b3349b6f3d098a89 : SUCCESS in 12m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e940e50b58064a4789d5824772242325 : SUCCESS in 7m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c16b6171d6d3428e9f6a008fbdf1a68a : SUCCESS in 12m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d21585cf24ef4cedb2a5c1b970f78fa9 : SUCCESS in 10m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ba6d1302d0934bff9a4abecb8b99f124 : SUCCESS in 13m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/95b6e97933884642a629510de68d1899 : SUCCESS in 15m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/56748d03664c4a6a99c10525cc15214f : SUCCESS in 11m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cce08a1eb2614f3b99792f70fbaf9ddf : SUCCESS in 7m 45s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a0c86d413c544b72bb78cb2e0a248f84 : SUCCESS in 32m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b1fb3fcdbf7646969b83b0d67e252e2c : SUCCESS in 13m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d884a123a57045cc8cad0cafa7b7691c : SUCCESS in 6m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6f3d990ac28c460183d7c93c0e130b58 : SUCCESS in 20m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6216ae2b5d4d49bdb3f3dec22eeb460e : SUCCESS in 7m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1f38edbc14784eac802d4a2a88c0b8f1 : SUCCESS in 9m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/139aa7668ff84b6ca2fdab298fb36181 : SUCCESS in 14m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b6b8045769749289adfa21061c30643 : SUCCESS in 53m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a34fea1245924ccdaf2a79dd7bc6b096 : SUCCESS in 1h 22m 45s","accounts_in_message":[],"_revision_number":1},{"id":"98d39211fb089708f935022ce3b528d903f5d72c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:44:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f7436bb55333708b1039865c420935b766effd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:46:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d1265928f74e7894dd8daf08110a9f : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a60ee0b6b4bfd4474ba56cc11df46e0c4da3f00","revisions":{"9a60ee0b6b4bfd4474ba56cc11df46e0c4da3f00":{"kind":"REWORK","_number":1,"created":"2021-05-27 11:13:26.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/793381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/793381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/793381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/793381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/793381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/793381/1"}}},"commit":{"parents":[{"commit":"a1fa18d92350eddfc1621e11fb567e07e97b5b53","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1fa18d92350eddfc1621e11fb567e07e97b5b53"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 11:13:24.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-27 11:13:24.000000000","tz":0},"subject":"update constraint for oslo.cache to new release 2.8.0","message":"update constraint for oslo.cache to new release 2.8.0\n\nmeta: version: 2.8.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I709018dc678544915458aa8cbad8582956d32958\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I3c562e90f1b690e1f6c559c01ac8e685c7623036\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a60ee0b6b4bfd4474ba56cc11df46e0c4da3f00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9a60ee0b6b4bfd4474ba56cc11df46e0c4da3f00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
