)]}'
{"id":"openstack%2Frequirements~793607","triplet_id":"openstack%2Frequirements~master~I667035d204c7d98b92076211f8048687a2e55cbc","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I667035d204c7d98b92076211f8048687a2e55cbc","subject":"update constraint for python-novaclient to new release 17.5.0","status":"MERGED","created":"2021-05-28 14:12:06.000000000","updated":"2021-05-28 19:26:21.000000000","submitted":"2021-05-28 19:25:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793607-new-release","meta_rev_id":"caff7b4be460983f1a37e5182c8ed11b42fb6576","_number":793607,"virtual_id_number":793607,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 19:25:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-28 17:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-28 17:59:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 14:28:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-28 16:01:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 17:59:24.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"bd2fbc1c7a281cca28afdeedaeeb39330f02b6df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-28 14:12:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa2d94d715be457cb04714cb237cf0855727221b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 14:28:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e0ec2d20188c444f9bfa4d15554ec8b9 : FAILURE in 12m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0ba23598eb73051f92dac3d5e5ec07b6ffec3171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 16:01:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f0d6e9dac6497a8774e768e90dff65 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a185e4fdc14bd2b8229400866e1726 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c209b4ec1e5141a186b2b21539de3311 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba9d095cce824d7d83710c8214aae004 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/643d69535cc54c529ecf6e7964e76171 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9b194e098b6342909a09213cc5c187fb : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5d2d1ef9f7ce4f37bc0d007b559121b1 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/19775a4b56844ef6863a977b8a9a44f1 : SUCCESS in 10m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/adb13ff2a90e4deab1b5ae4b3818d789 : SUCCESS in 11m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6fe683c58fff44228dfc127e867ba127 : SUCCESS in 11m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e72734f23e1e4208a30e6e5551193200 : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4c71be0f676040e1b308a24e95d912b3 : SUCCESS in 11m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0dd585ea3a3b4a0d85b9fb354b6a3b2f : SUCCESS in 9m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fe8c88a6f743492bb2e2622cf7008711 : SUCCESS in 11m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/686d5fb5ac834229acb95f283e4cd1ba : SUCCESS in 17m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8eb4964151f7485e83e8896bec4dcf7d : SUCCESS in 8m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1c9d72743ca34518b8cbeade45e794b6 : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/510bc04ffd58433f8511ac18780612fa : SUCCESS in 28m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a4108236aee747cc80184028dab350a8 : SUCCESS in 11m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3c3d62c63de24974810de3db5c23213d : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/241bb77b14e44d06a7d1dfced821deb0 : SUCCESS in 18m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f69990f2b30a40e48c3688efcd9f39c5 : SUCCESS in 6m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba695ac1ab3d4a0a92fd56dadb6cc46c : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/229a0db3eb9e4f65ae5fdf7252ff6d21 : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86107a3eb20745d89dd32a7215ea9d80 : SUCCESS in 48m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1bb8cce9a8e449ada64aeeca0c8957c0 : SUCCESS in 1h 43m 58s","accounts_in_message":[],"_revision_number":1},{"id":"8254628f98ead75aebe7ba7f9b86912e64d9f2be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-28 17:59:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d9662ed2a4015de5a6803ee43fc8b64b567052f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 17:59:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"facf447c064422a7a8071a28818b6abd14ba4919","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 19:25:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d4170d5afe24147b4293f398942bee0 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ad5a6be4604465b7fcaead85a1090a : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd936461a6664f3fb1d1ea45bbc4753b : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3b7d993dd7e401780ff172d55b0a3fb : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e222ffc87f047f09222a8be7f561dfd : SUCCESS in 5m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/632eb10285dd45fb97d7d118ff21a823 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f2388f5fc04e4268a16d63e4b13ebff1 : SUCCESS in 11m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e8730aac48f040cfbb0db78459063f7e : SUCCESS in 12m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/98e7068942344f52bb766532256c6353 : SUCCESS in 11m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6d385d5b738c476c9fa3a36f7580e90c : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b6b48cd17e674375a81bd4508a8d416d : SUCCESS in 11m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6a68de1b98894670a5a754de6e42f03d : SUCCESS in 10m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a83448e9ed7a4d6ab87d4a843ebcbfcf : SUCCESS in 11m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f69bcb1578404aec97948b795b189482 : SUCCESS in 11m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e7286824d3d24a84876804a2cf092345 : SUCCESS in 12m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ff63bfaf6f8443d383e7e1bbf4808829 : SUCCESS in 6m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a9683e8db5d64d5fadf72311c3dbfd27 : SUCCESS in 24m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/36e4458d9bb446738ef565ee9f8192a2 : SUCCESS in 13m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/260880da875d409b8c24ec08ab784928 : SUCCESS in 5m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cf00c7e1eccd45e09a50aea0cea1a871 : SUCCESS in 16m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6126db9c96e44afbb6bdf6d97dd3e90e : SUCCESS in 7m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2221a7729f9f4fd7a86fcc53f2d90cc4 : SUCCESS in 8m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f1d0dd91c533490e88d2d6b1cdf71ba7 : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3eb790155d234574bc6bee365dbfc476 : SUCCESS in 44m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec0af28a43be44919747bdece36c5789 : SUCCESS in 1h 24m 29s","accounts_in_message":[],"_revision_number":1},{"id":"10cb49b33264c3503b14d397880c0ea1a843a211","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 19:25:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"caff7b4be460983f1a37e5182c8ed11b42fb6576","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 19:26:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d11def2faac94ce19fdc1ff5bab77cd0 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a59e74ca0183ceb080ca1137a392cb09deb4a18d","revisions":{"a59e74ca0183ceb080ca1137a392cb09deb4a18d":{"kind":"REWORK","_number":1,"created":"2021-05-28 14:12:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/07/793607/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/07/793607/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/793607/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/793607/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/07/793607/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/07/793607/1"}}},"commit":{"parents":[{"commit":"55a5399c629bdfba82a39a0b6e3b0fff66dc2054","subject":"Merge \"update constraint for python-designateclient to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55a5399c629bdfba82a39a0b6e3b0fff66dc2054"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-28 14:12:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-28 14:12:04.000000000","tz":0},"subject":"update constraint for python-novaclient to new release 17.5.0","message":"update constraint for python-novaclient to new release 17.5.0\n\nmeta: version: 17.5.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I8f4aea2fcfbb3a7c52ae3c30a19ee18ca5f94234\nmeta: release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I667035d204c7d98b92076211f8048687a2e55cbc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a59e74ca0183ceb080ca1137a392cb09deb4a18d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a59e74ca0183ceb080ca1137a392cb09deb4a18d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
