)]}'
{"id":"openstack%2Frequirements~793608","triplet_id":"openstack%2Frequirements~master~Idf70c05f6bc57c2d6f6a89ef9e689d985744e593","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Idf70c05f6bc57c2d6f6a89ef9e689d985744e593","subject":"update constraint for os_vif to new release 2.5.0","status":"MERGED","created":"2021-05-28 14:19:45.000000000","updated":"2021-05-28 16:59:55.000000000","submitted":"2021-05-28 16:58:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793608-new-release","meta_rev_id":"18516e277af25cb0c7bc4dee2ad91aee00f4c792","_number":793608,"virtual_id_number":793608,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 16:58:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-05-28 15:32:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-05-28 15:32:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-28 14:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-28 15:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-28 15:32:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"22cabe06b84489907a256d2f2f5d878daf21ec23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-28 14:19:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a14cbbd3499891a2d04cfcfd7ab121f95824a3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 14:38:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bf862792685547d2818804d1cfb32298 : FAILURE in 15m 06s","accounts_in_message":[],"_revision_number":1},{"id":"ccbfd042f1d3b5a939a769db22d8119b79cc5a6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 15:31:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f53654ac42442b192c6da3e0f787821 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27b51ed94b25406f8f5657f520097d87 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdac63ebe6c24db7b7f3e170b8c82e97 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ab294124c654e3197bf39f435c403d1 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a33e8418d9545c4bc04cabdadb9a453 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/db4393f5e6bf446093dd3aa239ddf685 : SUCCESS in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ac99ec35a84942dcacb7098a7af10a41 : SUCCESS in 4m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5e7f8d19719a4b09b3afcc1c930d9ef0 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4a8b790d67ae452695528cce780c4cc6 : SUCCESS in 12m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8ff68250242341afb3902f85e2db76de : SUCCESS in 11m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/397ace548926406ca8ce28cdac10be30 : SUCCESS in 7m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/de4c24d167854870b40a7a75dac3bcce : SUCCESS in 10m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5b0c400cea564d5ba36ac993fdd8aef5 : SUCCESS in 13m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/c9db8b8348164dbf9b32314342dc46ff : SUCCESS in 11m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8a844805108445248805243bf9c73cc1 : SUCCESS in 21m 54s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2072f8cea1114e33b8e488e81169b4e4 : SUCCESS in 10m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9c802a3c415f433b9e7e670378f5fe3c : SUCCESS in 5m 14s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f3dd94eadc81463f98ee3133cf4d5b96 : SUCCESS in 23m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4614c235fb6247abb06672794154a2bf : SUCCESS in 14m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5d1147d3abea4a49be85a60bdb853f95 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a723d8a3d0943cd94e11cfbd4956d04 : SUCCESS in 17m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bd79571e3cbf449795a45830cd113c02 : SUCCESS in 6m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/487c38f888ed47efb491b310bab245be : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/87b253140c2c4551ba3a76865eb2fc4a : SUCCESS in 16m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/799449b3eed74d95aa67b9ac8b140eb3 : SUCCESS in 43m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0e90d08b8ec4587b491fbff92f0109d : SUCCESS in 1h 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"c08c4050482816c3bfb9e4d3920bd97fde41c09a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-05-28 15:32:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c84743c671fd14b5820e952e6638e427f5e95ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 15:32:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2b4c9dde6ddae0505a501fac66605964b9d7848","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 16:58:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/614f8665732541b1ac6d602f2ae1c257 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d38fcac61c4d9a9005b417dbfda365 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc5a9b2a74c24cc3b9cffd31e7c6af2f : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fd1db13705e4833a9692fbe0cef8caf : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32c704e3d6c34788a9378c6078243259 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c4da6ef663bc444e9c2efd9b1481ee45 : SUCCESS in 4m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/67f47c2b8ff248d9a3b0ae664dcbd45f : SUCCESS in 8m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ecac2b8fe72440528ad01a8498532cfe : SUCCESS in 11m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c5621411805f4ca7a17e5e84cd60e749 : SUCCESS in 9m 42s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/77fb2fbc1a5e4b049a1b41413fe6e3bb : SUCCESS in 7m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3dc9568362eb4ff486fe2b05ff10f030 : SUCCESS in 7m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7715e3fe05d148d1a7b84012e39482b7 : SUCCESS in 10m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/edcaa540b46743c19b3d3856f1f9ba4c : SUCCESS in 14m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/001e4ac629d04868a4ca79dc64d5bcad : SUCCESS in 20m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/18f7b3a4a2984c4da96999e48af3bc49 : SUCCESS in 16m 13s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/be4e51cb312248918be93216298761a0 : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b1b2c9cbe4874a2ab3e2a5552b8bb8df : SUCCESS in 19m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e12fe423674542ae80ab47a8f45c02cb : SUCCESS in 15m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d3d80a79608c4eb8bef0b94fd8e3d4d0 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/307764debd68404a96e4b82118e088eb : SUCCESS in 17m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/447b868cc16b484a8cf983bc52f097d9 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6f5be2c920614588ba7b8013ab6b21c2 : SUCCESS in 9m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7a757cd1c2de40f0a6ee1b6894adf15b : SUCCESS in 14m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d67a69d4902d4819ba2e157a89d6b415 : SUCCESS in 36m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfce924b81844ff789d22d4c8e300d56 : SUCCESS in 1h 25m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3f926ab75d268b7df4c840f8073c858e88b228ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 16:58:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18516e277af25cb0c7bc4dee2ad91aee00f4c792","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 16:59:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5f86b6735c4335a89ac5470673bf42 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1fb1490c487d7e8e36454d9ddda9220a1e82244","revisions":{"b1fb1490c487d7e8e36454d9ddda9220a1e82244":{"kind":"REWORK","_number":1,"created":"2021-05-28 14:19:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/793608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/08/793608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/793608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/793608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/08/793608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/08/793608/1"}}},"commit":{"parents":[{"commit":"55a5399c629bdfba82a39a0b6e3b0fff66dc2054","subject":"Merge \"update constraint for python-designateclient to new release 4.3.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55a5399c629bdfba82a39a0b6e3b0fff66dc2054"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-28 14:19:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-28 14:19:43.000000000","tz":0},"subject":"update constraint for os_vif to new release 2.5.0","message":"update constraint for os_vif to new release 2.5.0\n\nmeta: version: 2.5.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I28761067b2dba607cff21ade09e4bb58a09d10eb\nmeta: release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Idf70c05f6bc57c2d6f6a89ef9e689d985744e593\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1fb1490c487d7e8e36454d9ddda9220a1e82244"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1fb1490c487d7e8e36454d9ddda9220a1e82244"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
