)]}'
{"id":"openstack%2Frequirements~793756","triplet_id":"openstack%2Frequirements~master~Ia38dd932c3d8b5fea735b33dfb62e4444934b5b9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia38dd932c3d8b5fea735b33dfb62e4444934b5b9","subject":"update constraint for tosca-parser to new release 2.4.0","status":"MERGED","created":"2021-05-31 10:06:06.000000000","updated":"2021-06-01 06:48:45.000000000","submitted":"2021-06-01 06:48:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793756-new-release","meta_rev_id":"29230bd3d0a0825bd02d5c1faa3aed01c0d9dc3a","_number":793756,"virtual_id_number":793756,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 06:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 04:46:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-31 12:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 04:46:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 10:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 11:40:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-31 12:00:49.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-01 04:46:32.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d3071f7c0072f2aeb3488806d1a77055de451b54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-31 10:06:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"512f03767d695ff4c211b2f67a4f53c069b988b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:23:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/60bc6b6ab2dd49dda0437b31dfdb7012 : FAILURE in 16m 51s","accounts_in_message":[],"_revision_number":1},{"id":"10f114a50b0d3f939843bef8245ab26636c83b5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:40:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaa3ff178cbe46efac2dfb0e522bc8a9 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb52e055987d4d21bcb3a490b6e6f665 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a187f9f041ee44e1a033dc240977ee2a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98298463175a4122bd210fa8ffa76fb3 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3274dfd60164403a955fb7f6ed10de8f : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ff2b25e890bf4a0b9bda394edb9209fe : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2c936e00723343a0979dc2a8ce0bf1e3 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/b1a7b7e198314c07aa8d0dbaf3c73bbe : SUCCESS in 9m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c3096e3c20a54ea5ba455a352fbb6256 : SUCCESS in 9m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/65d8dd39524e48269f667e587ef746ee : SUCCESS in 9m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7b26267b57864621bb36c0c932cd1f7e : SUCCESS in 6m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ff3b66ba1aff4f39a8e6537d9597a143 : SUCCESS in 8m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/88445e13c033469ead2eb6bb19fb2b7d : SUCCESS in 8m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/522e430a537247bf8647f99221828514 : SUCCESS in 10m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/08142837a929402e90e0bbd02acdf298 : SUCCESS in 23m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d73a8197edb54750a72a9e67c61b4b85 : SUCCESS in 9m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d11f61b05613443084ed6685d6b5d3f2 : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d2f6fd9d87e8499f9c5914fa38f38aec : SUCCESS in 24m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9a5a945cfd5e4ad999130ad487451f8c : SUCCESS in 11m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/333053826eb84796a0f75479b3b3f27f : SUCCESS in 4m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dd66490d77794955a400e96e619f645b : SUCCESS in 19m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0b4f6920a60643a380f1513ab2ef3f97 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/262cc8d8e20140749084ef713f4d71b4 : SUCCESS in 6m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/46f4295a7f9449dc9d9b8ca78177febf : SUCCESS in 13m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45cc6ae27c394b4596fcadab3bbfd5fc : SUCCESS in 35m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97ef060ac93c44e5b1be9e6406a55a50 : SUCCESS in 1h 30m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cdb7cb94826d3280fea272876f252f60513569c7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 12:00:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c140e7fbdc868a3cebe5b919511ce71fee1f8622","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 04:46:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54738464503ada1b68459be89a45c60f91a9d708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 04:47:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42b736d5f6d02091aa48adcfb75710be26bd40c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:48:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef27d7bcf10d4e1d8f5451a13963c653 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f14af4c3ff14114b56918e0c9d02116 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8303e54bc0e34444ac21fce8456b9e3a : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a745c9d8c174b799e244a1d2d24c39e : SUCCESS in 6m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d66bfe4b943f4244b5eaab37e50fe816 : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/890d78bc54ce45348a9ac4cee1c570d4 : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cf283ff6e9654241988423c62f5d16e1 : SUCCESS in 9m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ef20f1166194e32bd42edda08b9327d : SUCCESS in 12m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/baeb1644415f4bee9cd67a7c96dc919b : SUCCESS in 9m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4977f5fe0e3240cf8d91e2338a086f9f : SUCCESS in 6m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a54172c9121c4f3da87c1bf9cd42778c : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/657053766bd14656ae6f4eac0aaa8e8c : SUCCESS in 10m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b00c6cd3f5ff4cfa87acc1b8a7c806c6 : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d18d5cd422f94cad9093c3a0afdcba88 : SUCCESS in 25m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1955f81811e346fba5122741c9b751cd : SUCCESS in 8m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0b3c531e4d594f33ad76d69c0a0fadf6 : SUCCESS in 6m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/861b64daf89a406785e3936ae48a2902 : SUCCESS in 29m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a6f6371ee13c49749b9b99e2488a53bd : SUCCESS in 15m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f9025bd800545fa8b4202f8caf06ca1 : SUCCESS in 5m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b6ca11ebd5724b589ef3b4c88e6d7e77 : SUCCESS in 19m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4b480c0ca4f0431f91743aa99fd1efa2 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9ab324a884b94f378bcd39ea4b277312 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8388b7e87d304af791aa82c05e49cdb8 : SUCCESS in 9m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28c4d12cfb444fa7a96ff2180b1ec796 : SUCCESS in 49m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a77738166b0040b288468ab158ea6a27 : SUCCESS in 1h 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"29230bd3d0a0825bd02d5c1faa3aed01c0d9dc3a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:48:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6004ade3b30faa43fb7bf449f4f782db3650e56f","revisions":{"6004ade3b30faa43fb7bf449f4f782db3650e56f":{"kind":"REWORK","_number":1,"created":"2021-05-31 10:06:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/793756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/793756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/793756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/793756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/793756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/793756/1"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 10:06:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 10:06:02.000000000","tz":0},"subject":"update constraint for tosca-parser to new release 2.4.0","message":"update constraint for tosca-parser to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I5a5aed8d4ff103af59e5e43e12d825416aff8e2c\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ia38dd932c3d8b5fea735b33dfb62e4444934b5b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6004ade3b30faa43fb7bf449f4f782db3650e56f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6004ade3b30faa43fb7bf449f4f782db3650e56f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
