)]}'
{"id":"openstack%2Frequirements~793757","triplet_id":"openstack%2Frequirements~master~I1826ee50bf7b90ed2645d3ab5f756b20be2e378d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1826ee50bf7b90ed2645d3ab5f756b20be2e378d","subject":"update constraint for heat-translator to new release 2.4.0","status":"MERGED","created":"2021-05-31 10:06:59.000000000","updated":"2021-06-01 06:49:04.000000000","submitted":"2021-06-01 06:49:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793757-new-release","meta_rev_id":"8e2d98b3e020b8dca0412f4e0650d193f2d4b32e","_number":793757,"virtual_id_number":793757,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 06:48:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 04:46:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-31 12:01:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 04:46:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 10:25:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 11:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-31 12:01:01.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-01 04:46:33.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"641c267a8f72aa02d13fae3e16eb504cd4ec8406","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-31 10:06:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c4e652e8b6a3246377a2ba9f6ea356fe4cdb9d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 10:25:46.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5771e8c77f2f4e2181676789174e6de2 : FAILURE in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"77f50e59beadfaa035613a2c30594e09afff7b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:56:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eb50e8ae55e4c479683a1aefa47f329 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e859e3e1854f7db3a29a5b0cbce4ca : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/357ae73982284c28b2a4f9d5eb498c70 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff52c36751a6442a9b73a0501a009535 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efd3d3f7a25d4b9fb1824bd6cbd11601 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be4d3401896b440180723048a60df84e : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48e97d44609740029e1b8db8ab1e1e8e : SUCCESS in 4m 08s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/1bb62101b5c744c6b7c5d521b94b6462 : SUCCESS in 11m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d54ae53500824f1fad7bddda2c8c60b7 : SUCCESS in 11m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a033b145bcfd4cfd8187d017b9702a66 : SUCCESS in 11m 59s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/991a7601e28a43128d37c884bcb5bec6 : SUCCESS in 6m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/03dab6f52a0f4dbeb7bdac85a9f3d817 : SUCCESS in 11m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/baed7b60c7df465e842d3139a5089c89 : SUCCESS in 9m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/222c944a25ff46b4bc81419ceb12fbe7 : SUCCESS in 11m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1ca12eaeeeda46b589c4e78fe5b20178 : SUCCESS in 12m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9647eb339c5f4a05ac89dd6f78866077 : SUCCESS in 9m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/040343bf864544f4ba72b7fc33f999f1 : SUCCESS in 6m 47s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/30c6609ecf10426b85aa2c5ff9eec229 : SUCCESS in 29m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/871c4545d3944169b896641d0c13abb8 : SUCCESS in 12m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c651adee482341f585541ffa2779f3f0 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6bfe4afd975340dcbd3909c2e682192d : SUCCESS in 17m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7845b12a98914d35a81a6be33738adf1 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a0bcdf18c716472ca0c0c4ca4d9279f1 : SUCCESS in 8m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/23d336d26cb04cecb38a2a2dc25a39d1 : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b4a5ad925f44030a5aa9ac621416c2f : SUCCESS in 42m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14377f24977a41f69001901efca92720 : SUCCESS in 1h 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2402ce9eb938d1079331792e47aaaa9a0fa5b540","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 12:01:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46d9f82d24bc0b715fcea4c05429d1a4eb7b2b7f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 04:46:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c8e6d3194235c7effe52510e4e2c3bd61500fc7a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 04:47:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"843d8742c7cfe51d98ec288088261dbf5b8918c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:48:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00edcdc6c5d74ca489ac3a06d092fc18 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/216f3a3cca4747adae0502bdfb94dc71 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1751cda921494841b5baf09767798383 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48951f68bebd4ddfb823e72a5f40a6c6 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be6124aff6d84933ba1571aa6da2ab1e : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bfa3f35d424b470cbc133f7ea8d5c9b3 : SUCCESS in 7m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/574fab9064b84a4e904703971c66d4cf : SUCCESS in 13m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/fb928db8e73f4625bfd1c32a66181207 : SUCCESS in 12m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/87226e6861d34fb2950e5b13495a8cdf : SUCCESS in 11m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f8030563c94a4ff39db5828a80ca872e : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a59d792c1004392a4097b2c504cfd0d : SUCCESS in 8m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6151d9f618b64f1b86c85b2b1e6df4f9 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/86293d0b933c40e2a6196a1e69519281 : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4be5d1364337486ea0a8d88a11b9f21d : SUCCESS in 22m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/83bd1ec964004c41a12aaf16808ad86b : SUCCESS in 8m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5d2cf1e8969413a971c27b3323d7b56 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/140e5fc506c44c2893f3eb8a009f200d : SUCCESS in 29m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aef1a4d891c4443980de5376c24785b0 : SUCCESS in 11m 06s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bad436a1059d48c99bbcc4e09c44f0cb : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/df2d06da74204891af0495d673f8db81 : SUCCESS in 20m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2a31a6256edf49059040c18156b8fa64 : SUCCESS in 8m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fe9db46cb2547398408ce01b9b49f9a : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d5ccf238d2b845d9aec39c9f4b38e813 : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce97a3502aba4df4823ec5ad51c428c5 : SUCCESS in 41m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e6b746493634e75bcf228c025b07e47 : SUCCESS in 1h 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"8e2d98b3e020b8dca0412f4e0650d193f2d4b32e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f9f5f7e30201ee63fcf6ae599d782e9789bc7ef","revisions":{"4f9f5f7e30201ee63fcf6ae599d782e9789bc7ef":{"kind":"REWORK","_number":1,"created":"2021-05-31 10:06:59.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/793757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/793757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/793757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/793757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/793757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/793757/1"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 10:06:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 10:06:57.000000000","tz":0},"subject":"update constraint for heat-translator to new release 2.4.0","message":"update constraint for heat-translator to new release 2.4.0\n\nmeta: version: 2.4.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I4d82603e13e2c6038276f60fcdedbd7d897c3b34\nmeta: release:Code-Review+1: Rico Lin \u003cricolin@ricolky.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I1826ee50bf7b90ed2645d3ab5f756b20be2e378d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f9f5f7e30201ee63fcf6ae599d782e9789bc7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4f9f5f7e30201ee63fcf6ae599d782e9789bc7ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
