)]}'
{"id":"openstack%2Frequirements~793764","triplet_id":"openstack%2Frequirements~master~I14f6222a407025e0de46e44159549531bdc9b211","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I14f6222a407025e0de46e44159549531bdc9b211","subject":"update constraint for ironic-lib to new release 4.7.1","status":"MERGED","created":"2021-05-31 11:02:31.000000000","updated":"2021-06-01 06:49:10.000000000","submitted":"2021-06-01 06:49:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793764-new-release","meta_rev_id":"11b84419699989c6bff53870027ff528e1edfb6e","_number":793764,"virtual_id_number":793764,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 06:49:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 04:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-31 12:01:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-31 15:09:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 04:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 11:26:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 12:01:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-31 12:22:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-31 15:09:03.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-06-01 04:46:34.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"770d1878c2c7a77c6a3bd737a059f67e0eba75b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-31 11:02:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24ccaf3820c6b0d279dbe0c61f50902b8dba74c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:26:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/53d03921448e48eb91343c6d13065080 : FAILURE in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3a479b916b47a301da1e2656dcab278ef637d7bd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 12:01:08.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f5a48cebe6beb8ee0f53b588fb1cd0211d014f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:22:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5935056da7264d0aa626c1ea93733597 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9f0f7dcbb91433daba8ae26a62824d0 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/021f1d5be30c4573b641d8f2ab72d970 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/380171ee2f494ec88a144c1cba1cbf8e : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a0405e45baf42309d7312e44453c57b : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a595969668534ab298356d8049d9a7eb : SUCCESS in 4m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c83019790a884e08a7ad21acc02ac354 : SUCCESS in 5m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9a73ba079e2417283e3f25d4aec451d : SUCCESS in 14m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/26a245e7b6244a13a1e792116563bdf1 : SUCCESS in 14m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b58f101476bb47bca2c008b53874d760 : SUCCESS in 9m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/37f2c0df35394a5a84fbb79ca33b0e67 : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ee2b5e31f010403f8c96006f21a69575 : SUCCESS in 9m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c55df84746674d288aa10ec1e5b1b19f : SUCCESS in 10m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a80e9fad339d4029beda686ae95aa432 : SUCCESS in 10m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6230b97059e849af830bbf493d814d31 : SUCCESS in 15m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0eb8381ae3f449859def050b851557c4 : SUCCESS in 9m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2552d5a7c92543d0bab873a293c2ae29 : SUCCESS in 6m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/718ecfb9f89240ad90331a90debe0867 : SUCCESS in 29m 29s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5349fc99a3214a17be29c6d11671c026 : SUCCESS in 17m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/833d458912ba4109b417086509afd444 : SUCCESS in 5m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4fed70da19df4618a515e2369929c26a : SUCCESS in 18m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7ccb6289fe834c0a91c08c05b2b6ff8d : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37c3d033c8dd458ab302b7525609eaf6 : SUCCESS in 6m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5a9ad205c72c49e88f17e717aeff1062 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48dd0ca38d8f41e7917e601b273db3dd : SUCCESS in 40m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7316bedcb4ec41f794d628b723601068 : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b610e17014360b9d2e7c7313a973d11a45b6408b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-05-31 15:09:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4fbf220ac7892389ce41b53cdeca04cf3708801d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 04:46:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fa264eee57c7f37b9499f53291a1bfe71c047ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 04:47:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ed80b775c11d7335841d06b4531ca1cef6e6bc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5972986dba4c3eb4abb6f52af35a20 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3963a1b95a194abdb7594c0b624de0af : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/589b88bce89a4f06950d54ee2561791c : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11c28f5e0bd24e58bd79d34c5ee502b0 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08d21be05a064e29a929bc5147d53dfb : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c911c77079e4134a1e0470ab1121d00 : SUCCESS in 4m 27s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/62209388412e4a6985770afeb0634477 : SUCCESS in 12m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/296b028940804255a8d28651d87f5362 : SUCCESS in 18m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ddd5e329bb914bd58a7644629221f6c4 : SUCCESS in 11m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/335c065defd5449097be4757c5be4b2e : SUCCESS in 7m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0a40ae984e744ef69aeb173e1963dd04 : SUCCESS in 9m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/42f465cd52624ccab14207e02ad30793 : SUCCESS in 8m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7134d521a5994c9fae518090e5d9ce10 : SUCCESS in 10m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/01fb256765674ad0b39583d567a0e284 : SUCCESS in 12m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4772dcf221b24ab39ff913832bc6b2d6 : SUCCESS in 8m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e0811c6bb02a4e6e8c9ab34721c71ff0 : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ff98498db1d4924833b897546df5874 : SUCCESS in 23m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5f9ae563f3b244778aa47ef41c2c3235 : SUCCESS in 10m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b3f36d38dc5c4c838a6831d9c90654bb : SUCCESS in 6m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eb04b75db45e4c6c9f8a15d3f299d377 : SUCCESS in 20m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/df59b86bd77d4c5b91a6c5d10c251a33 : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b84e1a5baffe47a691a43def4f9e091b : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ad52d6b17fe9485fb3269942c0e3ef47 : SUCCESS in 9m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af46a18569214899944a661393e91dbb : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a63f6bb102e347829964884eb30099c9 : SUCCESS in 1h 49m 35s","accounts_in_message":[],"_revision_number":1},{"id":"11b84419699989c6bff53870027ff528e1edfb6e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"acd0fcddbcf2a3203945a294c17c49af061213a2","revisions":{"acd0fcddbcf2a3203945a294c17c49af061213a2":{"kind":"REWORK","_number":1,"created":"2021-05-31 11:02:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/793764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/793764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/793764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/793764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/793764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/793764/1"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 11:02:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 11:02:28.000000000","tz":0},"subject":"update constraint for ironic-lib to new release 4.7.1","message":"update constraint for ironic-lib to new release 4.7.1\n\nmeta: version: 4.7.1\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Commit: Dmitry Tantsur \u003cdtantsur@protonmail.com\u003e\nmeta: release:Change-Id: I4d80d7d15cfc04dbc45b569b23ae55416be30384\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I14f6222a407025e0de46e44159549531bdc9b211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acd0fcddbcf2a3203945a294c17c49af061213a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acd0fcddbcf2a3203945a294c17c49af061213a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
