)]}'
{"id":"openstack%2Frequirements~793765","triplet_id":"openstack%2Frequirements~master~I4b1d3bf21b0e2efedfbd4c2da36359187420ec3b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4b1d3bf21b0e2efedfbd4c2da36359187420ec3b","subject":"update constraint for sushy to new release 3.9.0","status":"MERGED","created":"2021-05-31 11:04:40.000000000","updated":"2021-06-01 06:49:17.000000000","submitted":"2021-06-01 06:49:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793765-new-release","meta_rev_id":"d31bcc7b77d64bc446117b860fa7941ddabf717e","_number":793765,"virtual_id_number":793765,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 06:49:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 04:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-05-31 12:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-31 11:20:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 04:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 11:20:35.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-05-31 11:26:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 12:01:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-31 12:24:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 04:46:36.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"714ef14eec92b876e1fa2dc248c8541019e0ff6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-31 11:04:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"538df6c86a8e0299faac1d1d9586cea1d9d4af73","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-31 11:20:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"34f3b6426ce7b724da64222ad269c6f78c0bf4c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:26:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/49e9b17603ac4a748d1bf63e9c79f0a8 : FAILURE in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"68247114bd2357483dba9c689de43e5b6ac5159a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 12:01:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"60225dc779c51d74e08c414d3d5dac41f9c5dca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:24:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fa9e75d86e8440bb3759e920572bfce : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61799276048744ef833aba1d2617ebf6 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f446aaf0c13448f59c6cca405e388af9 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85ffe421581342ffb74e3d8fc8db6b95 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d64e319ef2364e368483f75c9cdc1c24 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f87d0159a663443aaa7a719718ea2d63 : SUCCESS in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5b53c0459bed4ab8be225b64d4564908 : SUCCESS in 4m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6b5310c56296464483d9e4949856910f : SUCCESS in 8m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9eb36c2a741b4eb59ba9485f19b746ae : SUCCESS in 8m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f7158fc82d7d4a6bb05ecdfa6d7fae46 : SUCCESS in 11m 16s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/570ca3727d07432d9d46e105ffa1fc0f : SUCCESS in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/29cab6804cc44a5e87d627e94c757fd3 : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9336e748fad34c848adb65ba9d53001f : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5dcdb1c2a31d451daecd0efc363dd3da : SUCCESS in 11m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b1f5a90021bd4c5b81274e8d7093f924 : SUCCESS in 11m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/40ddb43a6bbd46728422f070e3d98fc3 : SUCCESS in 8m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/06b3e95a45be4acc8326c7fa4fbba987 : SUCCESS in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/29977bb9eb314f83ae04f24b7b70c9a4 : SUCCESS in 29m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b5563c3977bd4601a9910075c3b62029 : SUCCESS in 17m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72435adc86f44f6b8d1de10e0d80e18e : SUCCESS in 3m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/117314347e5a4ac1948a1fe9d1d526ac : SUCCESS in 19m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a9493d77f5e7489dbb00c31be6791dad : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/87f4201b19bd4c06a0251f3be53663bf : SUCCESS in 6m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3d041cfe647745f48e6ed8bc629455a0 : SUCCESS in 9m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db983ced206a489a97440efcbe49725e : SUCCESS in 36m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db90a7b7a63449b1865805330ef3dd31 : SUCCESS in 1h 18m 03s","accounts_in_message":[],"_revision_number":1},{"id":"3e1b406f038c5e8b71ad391d9574065ce3789507","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 04:46:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"300f0b7936eb4d54d93449b1bccd86b5bf61e355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 04:47:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7efe83726eb73cbcff369c62ae146e5a97be7c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03835f304d394db4a44d29eb8d7422b0 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f05e3dc7d54749619b54a2dd92b46305 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d20958d006084863ae5f833920415565 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b3d25f9780b4326b7c35d2893a18745 : SUCCESS in 5m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4e51dc0c6fea40a9b1d5459848f0d45e : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/16a27e3b6dcb41b9bab1927d86daf181 : SUCCESS in 4m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/092e3df6c1204539b97305d241682ff8 : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21f90a41ce3d4105aafee33b37ea2b69 : SUCCESS in 11m 33s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8c40b8b94dae4103a9413151ccb8b444 : SUCCESS in 11m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3638d342d03e4011bcc85b668f574934 : SUCCESS in 6m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1e2f3b5f4308408ca56d0b011558a175 : SUCCESS in 12m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7d84fd4ef558451a9e6b33922188bc9b : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/09fc8d2c056143919510a05a7b2a035a : SUCCESS in 11m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/db483d62e0344748b1bad757cf94a2e5 : SUCCESS in 16m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/85963a070b12442b95b3df57bf4c606b : SUCCESS in 10m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6d03cffc74394506a69766cb6081f249 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d5f0de0d05424a2397a97ccdd032277a : SUCCESS in 23m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/24102bf793e6480689f8f8cf5e68201b : SUCCESS in 13m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f7b05c0e3cde46f0b6b5d3ecca921970 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/528ad5741eaf498f93b930d5f244e486 : SUCCESS in 18m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e17e71df55534b9b9e6d27e833306048 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/252c547f19994eec97bfe8dc6a448de1 : SUCCESS in 7m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be026fafb8bf4d8a8f6173bda08563ef : SUCCESS in 10m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f0cb8fc99824bfb914405aa84c58d8a : SUCCESS in 46m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c0b09214a7a4ea5b3e74a29e16a1c40 : SUCCESS in 1h 35m 17s","accounts_in_message":[],"_revision_number":1},{"id":"d31bcc7b77d64bc446117b860fa7941ddabf717e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c14bac1d838290fd012262e08a24ae0608fb172c","revisions":{"c14bac1d838290fd012262e08a24ae0608fb172c":{"kind":"REWORK","_number":1,"created":"2021-05-31 11:04:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/793765/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/793765/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/793765/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/793765/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/793765/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/793765/1"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 11:04:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 11:04:37.000000000","tz":0},"subject":"update constraint for sushy to new release 3.9.0","message":"update constraint for sushy to new release 3.9.0\n\nmeta: version: 3.9.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Commit: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Change-Id: I4635ee3805ec4894cf9314e5c3634c81b29b1b1c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4b1d3bf21b0e2efedfbd4c2da36359187420ec3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c14bac1d838290fd012262e08a24ae0608fb172c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c14bac1d838290fd012262e08a24ae0608fb172c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
