)]}'
{"id":"openstack%2Frequirements~793769","triplet_id":"openstack%2Frequirements~master~I1a607cbdfdc6f7c95d6359ba6ee3f38e1c6a940a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I1a607cbdfdc6f7c95d6359ba6ee3f38e1c6a940a","subject":"update constraint for neutron-lib to new release 2.12.0","status":"MERGED","created":"2021-05-31 11:13:22.000000000","updated":"2021-06-17 21:45:15.000000000","submitted":"2021-06-17 21:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793769-new-release","meta_rev_id":"639c42ae976365950ddc40dbff654aae44277b68","_number":793769,"virtual_id_number":793769,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-17 21:42:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2021-06-17 12:10:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2021-06-17 14:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-06-17 12:37:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2021-05-31 12:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2021-06-17 14:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 11:30:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 12:01:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-31 12:37:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 04:47:03.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2021-06-03 12:22:56.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"CC"},{"updated":"2021-06-11 16:46:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2021-06-17 06:01:56.000000000","updated_by":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"reviewer":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"state":"CC"},{"updated":"2021-06-17 12:10:29.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2021-06-17 12:10:54.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-06-17 12:37:27.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2021-06-17 14:46:10.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b9131166c6f0b979dfde5d90c51903ca1978bd9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-05-31 11:13:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f00368cc6651c2b4cffe00680a9c7ff0fd408753","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 11:30:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/496c7d02dc46495ba7c0d837b32539c3 : FAILURE in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"730897e80a0832d5ce707d8276c2764d1b18b746","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-05-31 12:01:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"91724f71e50f1aacf146032189abd14d8a639f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 12:37:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03d011802a8449d1807877a2e383b687 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f72248130fc644628da1ad72283c3374 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/71aaebb3352342f7a64cdaa4bacf45f6 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c165b7db87d94d6f93e037af6b6e12f3 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0b295ace59f433ca753f112704b79f5 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/027b471ca09745d793d051522a20f432 : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0dca9924fce540f7927ab93c514bf77a : SUCCESS in 5m 34s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/9de22aaee230452e92c2701239ee62dc : SUCCESS in 9m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/345231a1076d49eb8c45aa85d923e0cd : FAILURE in 17m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3af4631ad5cc4ad49619c862639066e7 : SUCCESS in 12m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c50156255fd945ab95052f60e47cde9b : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/474f668691f8493f9bf0e5aba777da84 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/989774cfacab4704bb78327017482df1 : SUCCESS in 10m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f1ec18501bf44866a59274a7b8565c91 : SUCCESS in 10m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9ed4384da64b49b6be89a11c9162422a : SUCCESS in 12m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/29eb9c6677b64ebbb8b2d445fb413cf6 : SUCCESS in 9m 51s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9314e920639b4eca8231c33c640b407c : SUCCESS in 6m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/54b26791cf6b4ec795779d81784eae9b : FAILURE in 5m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6b53d8912e8342b2b3716119a1156e84 : SUCCESS in 11m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9aa74ba91084460aa25ef5cfd5172a18 : FAILURE in 7m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b102c0aa70be49fd9ab4260f1574e6f8 : SUCCESS in 16m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/40db700f97e94797bf6e7fc8465807b4 : SUCCESS in 6m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7a958c47f024ef28e017d4fd2f5b349 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/faf9f756b7c64ba588969e53ac59aad1 : FAILURE in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/469e3d88aab948b4aefe264df20dc169 : SUCCESS in 1h 03m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e146dc3e32e40308b5fc66eb74b64e4 : SUCCESS in 1h 22m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0d0cf9ab6e0d92f8f3463080ada7046bccebca26","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 04:47:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"44ed338af15c52ff07b28c9847a76301bd4e0230","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 05:06:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f083d90a11944446afffe0f73d0746ed : FAILURE in 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"48c678e7e23ddec1be434845f9f0de4cfc6b5d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:03:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56168569e0a04483a004ddb116b1e77d : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/344d390f2d034ebd87675efacbc37cd4 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc17244cb9e84bd2b4b8028c4f585a26 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11408332856d44ce90c3adb824bff4a4 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7039fe5ea5f4dbb9659f0dba0dbd4e6 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f6a7bcbd98947b29b5a0a40cb15e270 : SUCCESS in 5m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ff8ed9c517f8412ca94ef95f47ac06fb : SUCCESS in 4m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/df957d981c89478198585b2e9442b8d0 : SUCCESS in 11m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4472e65d350c4f5f86d53c3038e705d6 : SUCCESS in 12m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1e6693c504fa4ee081d3a1efb8b9ebff : SUCCESS in 12m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/12d8feaf602645f28131b3135ed9a949 : SUCCESS in 6m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e3951d12dbc34b2f90909bceeeccaded : SUCCESS in 12m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/333182ef1352492986253e4d09a21fab : SUCCESS in 7m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2d7d93eb493d4cf9bd13962ce7b98c23 : SUCCESS in 10m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e18e5d3fa99f449289f7b064a2ae23a3 : SUCCESS in 15m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c0f66a7a571c4f0997d4eb8dc7fb04ee : SUCCESS in 12m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a7b1334760cf4c9e9d5e5301f0ccadfa : SUCCESS in 5m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9125e366d54e41d28cf11b8c8b9491fe : FAILURE in 26m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f7633e3d02c94c128e1ec0a7d4c91765 : SUCCESS in 9m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6b374f4f06604a11b65d92558cb80805 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/067772a4d4434616a47180dd4f10545a : SUCCESS in 22m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/43ac97448fd54c9c852ddbbf8e97ed4d : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ca7b6afa1ff64e2d8b69e85d8e516a52 : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bae641e72f1948e39e930825a57a5e21 : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ba36aaf5e1445f682ec8f477a1a56e8 : SUCCESS in 47m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d17a21fc827497c809964756b30a5f8 : SUCCESS in 1h 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1e642ef7ff477d7f354d65272e020a02a0caa5af","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 06:05:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cfa05831702e1b0c32bcca240f2df0c4dbdffa13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:29:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fb9cbc4211a942ae96259add4c917116 : FAILURE in 14m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c1dc9afecbdb220378166a0662d454df654fd652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 08:09:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce2051362c04144b33f0aa0d64efb93 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e999ab8e611147b193b670158fc7efeb : SUCCESS in 7m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d3b37c21b041425999635939f6f3b8cb : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f78e609dfab4b79a85667712c029259 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/725e623cb1f94e6ca57151bd2a0ee9a8 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14586331803b499d905d02ad25aaff0a : SUCCESS in 8m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/088ee48a25454adc9c220f1145bbd318 : SUCCESS in 8m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d01c2bed222b4f4688809b122eef4058 : SUCCESS in 11m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84d5ce99c8e74f678b1cb96e69cb1890 : SUCCESS in 12m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a552f2cf57fe4c059b81af7d1c0c515e : SUCCESS in 11m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3c07158f1ece4f22a8ad780ac328d75a : SUCCESS in 7m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c773fd7d6b8d4fab83cedef5d14a1630 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5d63818e80d3423e8bd9392d5397bc90 : SUCCESS in 11m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b0ddd4c0eb294a5381776b3771363d1b : SUCCESS in 12m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/79408cb4bee14cde9588201d5d026487 : SUCCESS in 15m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5bce39560f104cf98dfed3026b6fd485 : SUCCESS in 11m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/38ae6d3798e749ca806528602ca0bb9d : SUCCESS in 7m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d0746d49034142fda949c7d7baef5ea8 : FAILURE in 19m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/53e6accfbe92460ca5ff3cd3ab94d56f : SUCCESS in 12m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5290691ea7549a082ca1535f224918d : SUCCESS in 6m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/854e7063eb754c3483b4bc9d939db67a : SUCCESS in 17m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e55e6584e7074b698f9bd7f3caf17721 : SUCCESS in 7m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b678c1d9add14a3a92d829e996108341 : SUCCESS in 9m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2c6148d725994ff1856a301205cf2fd9 : SUCCESS in 14m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e822ed841644048a2ea8ad6783e16eb : SUCCESS in 48m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d879d29220d42ac9b13ac5eda63d852 : SUCCESS in 1h 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"701624978dbd43c3725c6d42bda88ecfeff029f0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-03 12:22:56.000000000","message":"Patch Set 1:\n\nNeutron job failure may be legitimate. May need someone from that team to take a look at this.","accounts_in_message":[],"_revision_number":1},{"id":"666f750625876da76793df3325a9414245796e66","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-11 16:46:01.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Neutron job failure may be legitimate. May need someone from that team to take a look at this.\n\nRight, this is addressed in https://review.opendev.org/c/openstack/neutron/+/794236","accounts_in_message":[],"_revision_number":1},{"id":"b5f9afc26775b140f851a4c1eb436534674aba1c","author":{"_account_id":9531,"name":"liuyulong","display_name":"LIU Yulong","email":"i@liuyulong.me","username":"LIU-Yulong"},"date":"2021-06-17 06:01:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78e74b2f9ef77f959f789c00d68a2b7fd5cd312c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 06:34:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/792b1dc183224796ba07b08328730fe4 : FAILURE in 27m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1c2e0d2bba9f837c14c9791facd063de49662be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 08:06:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faddfc1a7c6740ef9e267f1a2220eef5 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345ece79da0947349c71b7e955eebbfb : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f5a988ae8ea46e78a7c52aab56d72aa : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a1824dc4a4e4613870a52af7fda9f9a : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b1832738c1a45bca2b8d5592650ab3b : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/98564c0ec65c47be811be9afb7014aee : SUCCESS in 6m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/537e0b7e3d1842bc9d9998f33d11c150 : SUCCESS in 7m 46s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5571233d170b464ab5116b46e67b63f9 : SUCCESS in 13m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c1ba768f1ca4b9babc65317d34c1cb4 : SUCCESS in 14m 03s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4a4b3604deef4dfaa177e770196dab00 : SUCCESS in 11m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/fcf6bce11cda40d6a2bc7082e4ff7455 : SUCCESS in 7m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/11190dc30a2843f1b73eb10aa13cea54 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/444d068fd056428c8ca08422612889cc : SUCCESS in 10m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/15865b58c96c43ddb9fccff9d8dfee43 : SUCCESS in 12m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a48de76d75854bffb86676a703f6d792 : SUCCESS in 36m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6c34cc06efd34fc99bec1ab81ec8aba9 : SUCCESS in 10m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4758c60b0a17461396f563df757fe8e9 : SUCCESS in 6m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7d7ed4baf8d34795b61169d15fc6b61b : SUCCESS in 23m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aaa3451eb67546dfb0bb1b9b7a68961b : SUCCESS in 15m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/23ac58a35b0546a6858cfecf6941d5be : SUCCESS in 6m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/148385742f4940709897b7c1b6629de0 : SUCCESS in 21m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3dea5bfff14d4655b9025fb2afa4fff8 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1aa90754216e4ab39555770ec1e03cfe : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bce3ab8125a84f3b86c6d72664e0bdc5 : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d1d332e8c7340eeb79916645b189a72 : SUCCESS in 50m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5195563d2d8b44cebabd3b1f28f9dfa8 : SUCCESS in 1h 25m 10s","accounts_in_message":[],"_revision_number":1},{"id":"8aece0d350b6802373d5af2966e7539dfa336997","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-17 12:10:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"98bb81abfcafca6e92fe387215241e88b1586e2b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-06-17 12:37:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"439ee40f1eb8d35316f60966a7f1184e36886150","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-17 14:46:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"450eaa38f0b2598bb0007865e567882fffe8ba59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 14:46:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9363f64379b649e38bdf6a15202ad3594216793","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 16:37:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/795f0a2407734375aaa0fcedc55d2c41 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4eacb4910284cf489307c8377550d01 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc5b9f55662b4b3c89edbe854fa0ffe4 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67f005d34d2f4949a9f421fa7ade49a3 : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cea9a2f8f2d647eb8eb970c444c230ab : SUCCESS in 5m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4ee0422bed2470eabdb42259e6ba572 : SUCCESS in 6m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/944d169eea4c4691b9be3f194b6c25b1 : SUCCESS in 12m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f513f2d680684f55b9155e74c35f12b5 : SUCCESS in 15m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/eacc59ff353144e4b6d117be12a7df82 : SUCCESS in 15m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4254028b8485472dbb07c0268fd0991e : SUCCESS in 10m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4fea382ea6704295bb6cdd0dfd89fca0 : SUCCESS in 11m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/797e920395cc466fbf14272313d485db : SUCCESS in 12m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b8969889fece4af4b6ecceae9175d948 : SUCCESS in 12m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/95be29ce4492477ba5213612c769cd9b : SUCCESS in 22m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6b546c0ae7a244b2bc152c4f88c4427e : SUCCESS in 12m 36s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a94574b1143b4deaa9f1d8048bfdefb7 : SUCCESS in 9m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ad788e6a3f404683b1b87e173eaa94da : SUCCESS in 24m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/bed0d7922e944321ac5c3ba321e1317d : SUCCESS in 15m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/679b8940429942aebca93db76ce965e7 : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cb8ba33691574ae4877f67b471dcf12b : SUCCESS in 23m 32s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d21146ac9ec34096af01d83145c1e24c : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e2f66c3e554f46d19248114b750625d3 : SUCCESS in 8m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b97cc1fe02474f9eabe01f93f2dfb8d8 : SUCCESS in 19m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b99373dc001490fb40daca7f74f68a7 : SUCCESS in 52m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed0d62bbe4a141e98852220cc4c54e09 : LOST in 1h 20m 50s","accounts_in_message":[],"_revision_number":1},{"id":"fbaee718253ebd853bf90b38f0c5786957e3bfa4","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2021-06-17 16:50:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"08e7e20bba45ce9956077918e6a64ee9ba462120","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 17:10:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/24bf271f34674a888c9771035e73afdb : FAILURE in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"463c20f7ac452be7ef333ebf658bdb6b0b3e372c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:34:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62edb7b667aa4a2592b612da37dc2105 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1adfee27d44a4acb83c874299917a9ad : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abc9e0be587b41ddbf50bc6210b785ed : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a8da50829134926a3d03b06207f0895 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb37df66f7fb442fbe3f7e30c29c7a3a : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/15a75731d14246c4a673445d4109aeac : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f4721d56125e4a6a9b89057f6c1f92df : SUCCESS in 4m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dfef0e5220374ee29869aec3874f51c8 : SUCCESS in 11m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fc36fd7420e4ce5913722f6f63b26c1 : SUCCESS in 12m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fbc2b65ac7484624b6d8fba5538ec7d7 : SUCCESS in 6m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fe95e332946b41ddabe276bbfac20398 : SUCCESS in 10m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1d2d611fcb934525847cb35a6a334d64 : SUCCESS in 5m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/39f1d4d17df3487496d43bede6aa0a8c : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/819865779640408b97fd39166dc32f9f : SUCCESS in 8m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/fc9e3bd4b7ce4c5381645500038faf1e : SUCCESS in 11m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ead34ff33f1a4713a23cf83cbac0d76e : SUCCESS in 14m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/029b4228ce3643e9870ed501c78d2b6c : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9413f503d798439b969ed80c68b0741a : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/45d32930cca647edba6dfd5ea13e8b46 : SUCCESS in 25m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e4c2f3138b364834b5f134b74cf90270 : SUCCESS in 12m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a4bc79e01a554b5f9aa46818a0cd0f71 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5dc37b19fd854a839ca6cd3515f3bfdc : SUCCESS in 21m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d89fb4bff027408ea5238f4667a90f46 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff4d0244daa84c41863a888ddad2e58e : SUCCESS in 7m 16s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/31c1e02ab1ed472ea7045f8e3881ba7b : SUCCESS in 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fff82eab7234d03b80bb833f030ddc2 : SUCCESS in 55m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e64b70e4826147e19ced777a3aa3f2a6 : SUCCESS in 1h 36m 48s","accounts_in_message":[],"_revision_number":1},{"id":"44e4bf3648916e50f1d82a7ed1b962ad82c8ca59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 18:34:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc4d0ca19ea44b8959b5ea6b15e689664788b59c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 21:42:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5ae9d6e811e4ec9852ee1210fe24f68 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6cefbe0e7b845fbbc84cff0d7e748c7 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/234b04de1af4473593ac49b6eddccdd8 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/727e297fb2a04b61b39a7bc58f4f6b7b : SUCCESS in 5m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b966c41fcc24d869fae88d45c0c47eb : SUCCESS in 4m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/931e40a1afd74c8caee8aab828e72573 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5f4b32248bde4d8ebbf20c2f9d175caf : SUCCESS in 12m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/38a7e3adff614f66a45fc01b9cd92414 : SUCCESS in 12m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/440e9acd8741404f8d9a5f16a94a24ab : SUCCESS in 11m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/15124ca38eeb4ff8ba6e6bc48d228ae9 : SUCCESS in 6m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/442f21c68e91445f8faae61b4846181c : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/09103a370dc34a90b2bca90d44baa059 : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f26db1df4cb41439c9f21ed2c0bd4e4 : SUCCESS in 11m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5eb951c961004c8e891f67603521bbd3 : SUCCESS in 15m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fbea9845a6214fadbba4b53309ab5ae3 : SUCCESS in 13m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9d3a541e41464d928bd1b343aea3abd5 : SUCCESS in 6m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/29ebf9a6c44048eea7e3072ec6db0218 : SUCCESS in 25m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7382461b55b84539b9ddbf3c122540fa : SUCCESS in 11m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2f798e682f09443882bd8bfd14c9a5a3 : SUCCESS in 4m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c3068e838f05484e85b1b53665f99b6b : SUCCESS in 22m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f481adfcc3994466b1b22c467ebe20f5 : SUCCESS in 6m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c8cc9479f2c84397a4f5d2e0888067f1 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1357b8cba3944858bcc0175f9acf5770 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1dc2316212442499c7bab7bc00a290f : SUCCESS in 54m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd18e6e615114618b65dbbe116a9bcdc : SUCCESS in 1h 16m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4eae0336b74f75c3bc5bbde0d0b089d9127db1c2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 21:42:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"639c42ae976365950ddc40dbff654aae44277b68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-17 21:45:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a53bfb158c0b479cac455749bc4e811b : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"230fa080ebd33048afa575076e22e9f96480dafe","revisions":{"230fa080ebd33048afa575076e22e9f96480dafe":{"kind":"REWORK","_number":1,"created":"2021-05-31 11:13:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/793769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/793769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/793769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/793769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/793769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/793769/1"}}},"commit":{"parents":[{"commit":"d3eb093041ca83eed0f8eef68299e2c9b07c3453","subject":"Merge \"update werkzeug and Flask to 2.0.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d3eb093041ca83eed0f8eef68299e2c9b07c3453"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 11:13:19.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-05-31 11:13:19.000000000","tz":0},"subject":"update constraint for neutron-lib to new release 2.12.0","message":"update constraint for neutron-lib to new release 2.12.0\n\nmeta: version: 2.12.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Commit: Rodolfo Alonso Hernandez \u003cralonsoh@redhat.com\u003e\nmeta: release:Change-Id: Ib6fe6d7c8a507a312acad209071d34ab76f9f5c2\nmeta: release:Code-Review+1: Slawek Kaplonski \u003cskaplons@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I1a607cbdfdc6f7c95d6359ba6ee3f38e1c6a940a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/230fa080ebd33048afa575076e22e9f96480dafe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/230fa080ebd33048afa575076e22e9f96480dafe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
