)]}'
{"id":"openstack%2Frequirements~793797","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Iaaf636b6351a636f6b637ecdf3326cf1ea5cc271","project":"openstack/requirements","branch":"stable/wallaby","hashtags":[],"change_id":"Iaaf636b6351a636f6b637ecdf3326cf1ea5cc271","subject":"Bumping upper constraint for validations-libs","status":"MERGED","created":"2021-05-31 13:48:10.000000000","updated":"2021-06-01 06:49:27.000000000","submitted":"2021-06-01 06:49:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793797","meta_rev_id":"1967cad63711d6a3b6577155969468f31454142e","_number":793797,"virtual_id_number":793797,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 06:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 04:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-05-31 13:59:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":1,"date":"2021-05-31 14:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 04:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-05-31 13:50:30.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-05-31 14:09:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-31 15:13:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 04:46:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ce65ed0e1335fca47e0700a7ee473a19d4c6c7b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-31 13:48:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c5aff33da63a5c188b29da41892bdb6c516e71d","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-31 13:59:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7e43ee83ca6c21794e1dcc7436e2b9d9adac5572","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:09:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0e975b4c5514494eb775ac297bb0420d : FAILURE in 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c43eb4ccf0ebb0b2e54f91dfd303bc69c1f6e278","author":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"date":"2021-05-31 14:45:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5b383bb10146242fc7a02109424bfb9d42a3a917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 15:13:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6625e88a296a41ba90e748d880a078c2 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f93d8d393bb34813a92fd6eb2c9bc0d7 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/507d496eb16846a0b37682354c8f8619 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afe9e30f7d0a49ccb28eb95fd87b43a4 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a98d32111c884ac5b6a946464fe90d25 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c15cf4cf03f43bd9e9f492d815e0548 : SUCCESS in 4m 15s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d54499820d904988871b9e801389c88d : SUCCESS in 9m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3cd99a01b6444ab1ae40baf571723029 : SUCCESS in 11m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/44ffe075f5284398a084028937396292 : SUCCESS in 11m 06s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6ff46aee81f3400db4186df82f60b5ff : SUCCESS in 6m 38s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/530c34d6dc4343daa85b102c47eff302 : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fb75df6b111d451a9384eb17dcbb148d : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7646fbb008b64c5a91ed728c1f48d46d : SUCCESS in 10m 17s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/0c85dfd1d23448c8ae31465c2ae9819f : SUCCESS in 56m 50s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/d744ed87c0a648618ad06989ce5b9521 : SUCCESS in 5m 55s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a6fb3348df9a4b1e8ac980a717f86021 : TIMED_OUT in 43m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/60485b3abf7049b0911693928fbb362d : SUCCESS in 14m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/30d35fdaffe74a69b522a7b2fd4afc7d : SUCCESS in 5m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/74ac6630e3b24c10ba6df0641bd109f0 : SUCCESS in 18m 10s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/d5d92f7b06424e68ac39736d9e9e3076 : SUCCESS in 5m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/76c41537e65a4087a082e3d9fc60de1c : SUCCESS in 9m 31s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/89833d23e31b46f28d5bbdf052431265 : TIMED_OUT in 41m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67b53c927c53478daf836fd6f20c1737 : SUCCESS in 51m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec7d3a9b157e42ad823cd923e7019c39 : SUCCESS in 1h 21m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b9804fe3ccfec2dd040bf1c672af2638d0f750d2","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-31 15:24:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b47724e80522d41a3a11231222fa459e96844e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 15:45:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/cbee0767a2cb4c60aff998d9abd9da4c : FAILURE in 18m 48s","accounts_in_message":[],"_revision_number":1},{"id":"7143559dd5bad9d1e525341c45136539f3749a20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 16:57:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34aab146175c4677903c196de6626d34 : SUCCESS in 7m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56b18bdcd03942adb392f24880659b32 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa116f1aeb24a778650c4ba3a9c0ee4 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98fe25e576904539baba98ffc848cb9f : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ed684b2c0e4435db97515880ac2ae3a : SUCCESS in 7m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc5eb358fbcb47ee965ef6d5c42fbbb1 : SUCCESS in 7m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/68eab854bae84eb0a0a5c077fb226a72 : SUCCESS in 11m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1b7c616c5e104090b1a51bb70457b2f6 : SUCCESS in 10m 00s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8c05a08d2e5446828d69d646155c7cc2 : SUCCESS in 10m 01s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/156b39200e334d0b97172767d13d7583 : SUCCESS in 5m 47s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/30fc76e1d11f401c936ff24c4f32460f : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f7f175fa7a794c418406b92fd31ad975 : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/030cf817914d4a3ba9a3337b3e206e3f : SUCCESS in 11m 21s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/dd171385766a4de2b68943f42835933b : SUCCESS in 14m 10s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/321354dbfe4f4ca7a4962c910fd66cc6 : SUCCESS in 8m 11s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cabff346bef44cde803ae1562f23f318 : SUCCESS in 33m 43s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/ae6a85eb71034fc4b05583eab409f2d2 : SUCCESS in 12m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/04fb007a11914f43b561c908dd042d5a : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c04d94e6c6d4de1b9761038c9c85b33 : SUCCESS in 17m 25s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/6e6e992a70a5421f9cbbfc94479b8f8d : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6ae68b4a2b947e2bf4d1948db283a73 : SUCCESS in 6m 50s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/a1e9ad54a83d4f349d58c98053d5a630 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29080af967b14a138febfe2c5c7279e6 : SUCCESS in 56m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fe13384cab74fcabd251f5752ca5a6d : SUCCESS in 1h 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e72250b49749c5ae33d851267f0dad64da3bf836","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 04:46:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"776013201ffbff1148121fd97213736fe515e61b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 04:47:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8b0874a7b47f703a9ed6c6b6ba1dc151654a8c6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19945c84bc4470dbe14f76d2c96474c : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bbf978dbdcf4d48b9f987626d77db62 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf6d590673834b71b7c1c612c90d246a : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/573ebd0b83cb4a928e718d7e7954ec06 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fe70f531c96f4e908087e558affc2cb5 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/72e4a59083f04c3caf24e736c96d0dcc : SUCCESS in 13m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/279aa77505fb42a892db29b9f787b6a6 : SUCCESS in 9m 48s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/588d9878772d43d0994acb48660d7ac5 : SUCCESS in 10m 49s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/2947534fc3e0485cb15cb17925453a2a : SUCCESS in 5m 29s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/c93acf908ebd4308903b9ddf19816344 : SUCCESS in 9m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f5bab1837654535886fedbc1b0b0d57 : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/855bec66ffbf425882531eb2400f11d3 : SUCCESS in 11m 11s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e775b27fde8c49c8a59fdad39bfbeaea : SUCCESS in 17m 30s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/ccc9f85fb778448fb7b5a1c921c33b55 : SUCCESS in 5m 40s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/1680ceaa044049cb9ae7d970f1e126f7 : SUCCESS in 23m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/2c0f06c2a4fe457a904cf504f3c8ad41 : SUCCESS in 12m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9d3b638ce0a743d3aec46027e11c0ce1 : SUCCESS in 3m 54s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/028d605c85f54a26b7e25b37304e7065 : SUCCESS in 17m 30s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c663626de268417ebf5ced3eed776fa3 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2a7996f81174a7dbf1519a0b2c1ff52 : SUCCESS in 7m 18s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/0651e6e5f824442cae6d055ce532b991 : SUCCESS in 9m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7219f4610b3a4cb689c2b6cbd84e9b0f : SUCCESS in 51m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2d98fa6e8fd940099deac4369fdc08c5 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1967cad63711d6a3b6577155969468f31454142e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 06:49:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4da5cd69bd025f1768bde5147d1b0c9974643ab9","revisions":{"4da5cd69bd025f1768bde5147d1b0c9974643ab9":{"kind":"REWORK","_number":1,"created":"2021-05-31 13:48:10.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/97/793797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/793797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/793797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/793797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/793797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/793797/1"}}},"commit":{"parents":[{"commit":"6522c8905d754ab396da2f61066011fcc3f7fece","subject":"update constraint for python-magnumclient to new release 3.4.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6522c8905d754ab396da2f61066011fcc3f7fece"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-31 13:45:13.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-31 13:45:22.000000000","tz":120},"subject":"Bumping upper constraint for validations-libs","message":"Bumping upper constraint for validations-libs\n\nNew VF CLI was backported to wallaby but the counterpart\nchange wasn\u0027t merged in the python-tripleoclient before the split.\n\nThis patch changes the version of upper constraint so that the\nrelavant backport could be merged.\n\nRelated-To: I65fdbd0ffd42411196b0b3a25690670b76dda2c8\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iaaf636b6351a636f6b637ecdf3326cf1ea5cc271\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4da5cd69bd025f1768bde5147d1b0c9974643ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4da5cd69bd025f1768bde5147d1b0c9974643ab9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
