)]}'
{"id":"openstack%2Frequirements~793900","triplet_id":"openstack%2Frequirements~master~Ic23689c9a5fa91ea0933d73dc3e4c822f9f3670c","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ic23689c9a5fa91ea0933d73dc3e4c822f9f3670c","subject":"update constraint for keystonemiddleware to new release 9.3.0","status":"MERGED","created":"2021-06-01 08:29:20.000000000","updated":"2021-06-01 15:07:46.000000000","submitted":"2021-06-01 15:03:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793900-new-release","meta_rev_id":"e3f99312ccc3634097421eff189e0c9081932291","_number":793900,"virtual_id_number":793900,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 15:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 13:26:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-01 15:03:52.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 13:26:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-01 08:46:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-01 10:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 13:26:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-06-01 15:03:52.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"b850ba9cecb9ad766de77dd3b0921283f070075c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-01 08:29:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"814d3e89f58e0ba802e91c9a5d36a3369d1a556e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 08:46:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1fbaa696bdb0477dbc4ae5b72ecfaf0b : FAILURE in 13m 21s","accounts_in_message":[],"_revision_number":1},{"id":"a2cd4416c5e3370ea68fbc2e90cc84463051bc0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:16:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56919070a0e94dd483b1ead5f5b35948 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b657db218a64c3fbb9d2c8f95ba54f5 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a34bbf9d66304b75b7743dcbb8aef183 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8085e306dc584336aaa3e58c0d69ecc4 : SUCCESS in 10m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67dee5645c1b426cbba5122af238e29b : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d5e85efc15e04d0b879245e4962e8ec8 : SUCCESS in 4m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/87ac8abd7cf5457f915dadffb37737fb : SUCCESS in 4m 12s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ddaf5cf5fa5e4872ad6d8b242ff53b18 : SUCCESS in 28m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/711fca68fbae46149625bd205e98c05a : SUCCESS in 12m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/52d26c8533ef42dfadd6593d0ed1a4fe : SUCCESS in 9m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/01bffc81de4942909e89b15beee16fb1 : SUCCESS in 5m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2c88c3230bf94258b74fcec8bdb93053 : SUCCESS in 11m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/80fe758c9274440db291fd9e03a1fc60 : SUCCESS in 8m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4899dbbfa3a84a58b2fa838183ee8fc2 : SUCCESS in 11m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/32107d4ed7ae4d0e929749ef1785e6e2 : SUCCESS in 14m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/68e9b93be5ce4c2cbc2ca816a68f38e6 : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a74f04759e2542cbabc0a7c90cc0d436 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/167f09dd22c746588802020985a38ece : SUCCESS in 26m 49s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ea397980127a42608fcaa92faf888a2b : SUCCESS in 12m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c422d617a204fabb3cc1b6648ff822f : SUCCESS in 5m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bcc56109ca6f4b94803c21c642979524 : SUCCESS in 16m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6fb723746a6945788e3a95c10cab20c8 : SUCCESS in 7m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d48d312954f64f17b90494e3bb334e44 : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/084d4322f929429c8ae1101f3c35233b : SUCCESS in 13m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32c65b08812e46d58d7a76dc8188bdc3 : SUCCESS in 47m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3abfabb429645e4ac2dd7c4a2efb007 : SUCCESS in 1h 45m 52s","accounts_in_message":[],"_revision_number":1},{"id":"07898520b8991710cc4a0bd9332aaad696b47f99","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 13:26:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f58ebb15acb83727cd1c845b3969ff945382946c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:26:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a798386ea6fe82cc15f4e438ce71490c239970de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:03:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0767dc91074713ad2f32d52ad237a7 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6620a9594ac248bfb6d0238202b00625 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f5414e93b544fe799b945d7acd95d00 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4d4fd5a0d5e4e9cb86936a48e120684 : SUCCESS in 6m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f1ab526e344d4fcf92a2884a16de8927 : SUCCESS in 5m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8a676ea16f3a49db90d7e2d5dabdb4b1 : SUCCESS in 7m 39s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/178bc5385d2d442f8e373041e312c43e : SUCCESS in 12m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b12fc7097549484ea836a2868e936102 : SUCCESS in 11m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c7f5e671c15b4409a147d371144ad102 : SUCCESS in 12m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/18a3fb78c6ae4e24a53c7ce5f0ce38ff : SUCCESS in 7m 14s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ba7649b09a304ff79abbe82de66b8ae3 : SUCCESS in 11m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3a9d726f1a03479695be1f968bdcdd3f : SUCCESS in 8m 32s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ef99e082fc9a4d33a6f69294527b4205 : SUCCESS in 13m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/34c2ce3bd41d4d748ba825060f0fcba2 : SUCCESS in 16m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7b90061534b0416d92383fb1d833b4f3 : SUCCESS in 11m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f345952551d4aa6879fe9761ead7398 : SUCCESS in 6m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c4ab077b3d2f40a09b67192be37618f5 : SUCCESS in 23m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f044cabd04c647998276c6b8b3e5dad8 : SUCCESS in 15m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/639ea00626e146a6b441b546908f759e : SUCCESS in 5m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/019b3f378c564c3f9f8fe5774083a146 : SUCCESS in 21m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5567ccce015e4a4b9041e3daa1b4eda3 : SUCCESS in 6m 41s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6935efbe04584063bb833bc2cd13567d : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/352131e36d4e45a8ba2f51814813392e : SUCCESS in 12m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b58683b73cc413789c6bce83d4704a1 : SUCCESS in 52m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fd76989dd624b0a9c68694789bd9cec : SUCCESS in 1h 22m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e357d515c100f9b52ed7ab7f2ca756f725ad311a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:03:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"07e09532bd170a167cee28add19ad70ab3e41e0d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-01 15:03:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3f99312ccc3634097421eff189e0c9081932291","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:07:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec7155fe5f645c18e1e1929982132f9 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"094589f1c2b58b4414660d7569f89582ff7e1a39","revisions":{"094589f1c2b58b4414660d7569f89582ff7e1a39":{"kind":"REWORK","_number":1,"created":"2021-06-01 08:29:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/00/793900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/00/793900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/793900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/793900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/00/793900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/00/793900/1"}}},"commit":{"parents":[{"commit":"2623351d90409a2e147025eae3e0d57c6de52135","subject":"Merge \"update constraint for python-swiftclient to new release 3.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2623351d90409a2e147025eae3e0d57c6de52135"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-01 08:29:18.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-01 08:29:18.000000000","tz":0},"subject":"update constraint for keystonemiddleware to new release 9.3.0","message":"update constraint for keystonemiddleware to new release 9.3.0\n\nmeta: version: 9.3.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Id44fd8e7c6fb6de734ad672d4277586f645d9de1\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ic23689c9a5fa91ea0933d73dc3e4c822f9f3670c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/094589f1c2b58b4414660d7569f89582ff7e1a39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/094589f1c2b58b4414660d7569f89582ff7e1a39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
