)]}'
{"id":"openstack%2Frequirements~793903","triplet_id":"openstack%2Frequirements~master~I520a6b029b8b98ee5235d2261d2cb3378adc7fb3","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I520a6b029b8b98ee5235d2261d2cb3378adc7fb3","subject":"update constraint for osc-placement to new release 3.0.0","status":"MERGED","created":"2021-06-01 08:42:52.000000000","updated":"2021-06-01 15:17:46.000000000","submitted":"2021-06-01 15:15:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"793903-new-release","meta_rev_id":"2f237add2cae255906e0df4e7e02bdfec99792e9","_number":793903,"virtual_id_number":793903,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-01 15:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-01 13:26:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-01 13:26:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-01 09:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-01 10:24:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-01 13:26:01.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"4ecb5714df096399a46a12980e063a4e8ed1ae6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-01 08:42:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43b37a446ca4ff255518801543a0f50b91a0b4bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 09:02:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4562db84a8f14c00b049898fd7594fac : FAILURE in 16m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0c040f75bfb3d7c7529ba029dfec3fdc62256578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 10:24:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f037a711203480698e3b3227c0006a7 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6e62f2bb474f6e96dd4109628a573b : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48d3c7da7f054c508748264f6b5537f7 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67bd77d4ce8146f982a336eb6b08d9ae : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e9fa778c33e4dd181e29d197aaaee4f : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/102d877e0e804d988d82e17d9ec99669 : SUCCESS in 4m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ad9a9d797be14f75a172b4d097c1d58e : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0a747ecabeb84cd794595227ebd41fed : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9682fce96b34ce0835e28c4259c6fcd : SUCCESS in 10m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3af708d3646445c4b82b9a4514f95fcd : SUCCESS in 11m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b2d4d792bd1047868e00020e879b48be : SUCCESS in 6m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8195ffe4abfa423498b2a38e82b86316 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0a02fb5768642c9a766551c738ce2d4 : SUCCESS in 8m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7e54c19922f242049053186adc2c3bdb : SUCCESS in 11m 28s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7cde61feff50422d8179877dce2995bf : SUCCESS in 14m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bc374bcda1294b2883bab19f18b3b118 : SUCCESS in 9m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bbe0fc90ffc4425fb43404bb6ac46c4f : SUCCESS in 5m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2f574fc07938412a9e1636af0da503bb : SUCCESS in 26m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9c88d14f56bc496ca529baaa5cf40f2e : SUCCESS in 13m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f89f45cc30e944fc8838c9604cde12e4 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49350604f92b4682bf0ce80b7d9cc71f : SUCCESS in 18m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fd476a217b814f1da1171293d0321512 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d8926e2f14c3425a97602f76d50e4119 : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/86bcbe54f22a42038e417b87ce78c4eb : SUCCESS in 12m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2149286f0da546a9b510fbfc671d3d0c : SUCCESS in 50m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f36b9230fcf4e5bb6a277b36f3ca155 : SUCCESS in 1h 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4b461caf1cfbeaaae1b3adda0e290605cdb2d67f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-01 13:26:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9d903f802883dffd18d12052c3ac565cdf757da3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 13:26:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2c3ec2c28cc8184ce828e6e95d60275eaed04a8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:15:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3b437eff52426997b54822e1d28426 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/893ef9c51fb34d19b929195919495850 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e55e2abff4d44b02878bce63bfd47e31 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95504b79a962479981330215aad843ef : SUCCESS in 5m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45dbc6acd8d34fab8000c566c94eed3a : SUCCESS in 6m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f8f435d9adfa4971906b0ecd3bde540f : SUCCESS in 6m 09s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/2f5b71d367314176b5c1115dfb8c68cb : SUCCESS in 11m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a424dccf617e4435bd4a3d71df1d52e4 : SUCCESS in 10m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ebea783c42714ecd9934dfbc1fe0d08f : SUCCESS in 11m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6683de396f35479fa1c7c171f42e7110 : SUCCESS in 6m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/26fb306d1d1f43ae8c1ecb30a5ee3e4b : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b13bb0d43d524942a24cf9f4b03564d7 : SUCCESS in 16m 31s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/16872df2dfbf46f7ae1acf17a5b022eb : SUCCESS in 11m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eed8e867879e44c5b0fa46e92f093ca8 : SUCCESS in 15m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9e91802542424f22bfa288470e30e389 : SUCCESS in 15m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8ec6a784810048c4aa32553c79c648a4 : SUCCESS in 6m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ebdb293b666d4dd09af7a1e96b4198ac : SUCCESS in 26m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6295f190cb7c4d94afa861c7a1548f84 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6a379d4772054464b06617203bd4a54b : SUCCESS in 5m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/308a0d6285934da7a0eb46f737440501 : SUCCESS in 17m 36s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3ac00cbf7efc44ed8389a17bf270534f : SUCCESS in 7m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cb1694f2064e40c6ae752cadd1cd3395 : SUCCESS in 11m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d453ac21e789495598cced6d1146e35d : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0697a36d9db445199997a9dbf2d9b264 : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/292c4ddb49634bce86b05320c2685b99 : SUCCESS in 1h 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4189b0ab2af0c9c9f6b04c8e43e7e49c7885d3eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:15:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2f237add2cae255906e0df4e7e02bdfec99792e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-01 15:17:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277ecf7b72b44ffda97efd0d7ac67d4b : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e719b0f2594650772dd9636656d67be5d173d6d9","revisions":{"e719b0f2594650772dd9636656d67be5d173d6d9":{"kind":"REWORK","_number":1,"created":"2021-06-01 08:42:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/793903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/793903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/793903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/793903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/793903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/793903/1"}}},"commit":{"parents":[{"commit":"2623351d90409a2e147025eae3e0d57c6de52135","subject":"Merge \"update constraint for python-swiftclient to new release 3.12.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2623351d90409a2e147025eae3e0d57c6de52135"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-01 08:42:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-01 08:42:50.000000000","tz":0},"subject":"update constraint for osc-placement to new release 3.0.0","message":"update constraint for osc-placement to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: Ib3650b06afb0e59d8473c0475c8fac5307ffb42b\nmeta: release:Code-Review+1: Balazs Gibizer \u003cbalazs.gibizer@est.tech\u003e\nmeta: release:Code-Review+1: Sylvain Bauza \u003csbauza@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Tetsuro Nakamura \u003ctetsuro.nakamura.bc@hco.ntt.co.jp\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I520a6b029b8b98ee5235d2261d2cb3378adc7fb3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e719b0f2594650772dd9636656d67be5d173d6d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e719b0f2594650772dd9636656d67be5d173d6d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
