)]}'
{"id":"openstack%2Frequirements~794260","triplet_id":"openstack%2Frequirements~master~I728024375b4d4d951251872a8e6b5977f59e710a","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I728024375b4d4d951251872a8e6b5977f59e710a","subject":"update constraint for openstackdocstheme to new release 2.3.0","status":"MERGED","created":"2021-06-02 15:54:08.000000000","updated":"2021-06-03 01:51:11.000000000","submitted":"2021-06-03 01:50:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794260-new-release","meta_rev_id":"162c2b642b8360e0eb00281b98ef33c11036f4e4","_number":794260,"virtual_id_number":794260,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 01:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-02 21:00:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-02 21:00:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 16:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-02 19:18:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-02 21:00:30.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"90d69ecadf0564f9a35f5e0f162df1bf5219cc71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-02 15:54:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce2e0f39538cb415fe71d05dce81971ee134e33f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 16:19:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/34e8120efc4742869de55b69e0ef6f96 : FAILURE in 16m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cfa1711039553b83edc9da34f781d7fb4239b250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 19:18:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161bfacb8a4d45a6801fb2a0ef442d63 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cede4268f0e848c0ba14672c05e7b4c9 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d23529eadd548e48339af5fc6c927e0 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc1e34838764748a02d0babcc9c1bec : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81c025dd63e84130ab0147362750fd0c : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9299d1af001426cad1ee32d07097bbe : SUCCESS in 4m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7b6c920d8ac949c9a3933407b420fff8 : SUCCESS in 5m 51s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/dc78c7c326164b1aae32ac03351fc4ea : SUCCESS in 13m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8365bc8d51974bf6809db06671c91178 : SUCCESS in 13m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3e888269994449f2b95ff16983d52e0f : SUCCESS in 12m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/062ffea1183d4f899b709a0d3bed21fa : SUCCESS in 7m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c75ddb155e11496d847b03dc3e097796 : SUCCESS in 11m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3934f1aeb4524b4e85cb747b8b1b10f1 : SUCCESS in 10m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1740ca28a0554594b024828c4742410e : SUCCESS in 12m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/637a45f93f1b4e699c91bd52644a33d0 : SUCCESS in 15m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/65147b0196064d21b9e7382a2ac6f1c3 : SUCCESS in 10m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/56e692e7e2ff4ebb81db7bb58cb0a3d0 : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/68ce1f3e27914e35b9ee9b51b04fa9fd : SUCCESS in 30m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8d72e7d1dd66401d930eac14cc45a015 : SUCCESS in 14m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5c0b1a5bc80b4385822a11ea186f8f75 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b8dd2fe095e54b5a93d516104bb69838 : SUCCESS in 20m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3704fdbf175749d5a8b2f848b8ec3b07 : SUCCESS in 6m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/741899ea4783412ebb75411d21f571ca : SUCCESS in 7m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ff86cd4763fe4d699607f95d09de99e2 : SUCCESS in 13m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/006eed405b054ada93b9502124fa8354 : SUCCESS in 55m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd5e36cb17824000aea8c22e7636d9f5 : SUCCESS in 1h 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"f5412298402f8a48ebb5683051e39529094667f0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-02 21:00:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fde714f1e4846ce9ae560695a2c671c36ee58b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 21:00:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7850e4e4bc3c746e6f3e73835d6ab8c92d001a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 22:56:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf61a7ce179d4887bfecd65dd80dee2b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90a7cf82488e44879f5f80eaa6682457 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58c50b9de9534db5900da2f0e4a51cb5 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/059fa464d2f84c17a433d33faf9b86c2 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/08f2567bac7c472e842c0923e19c44b7 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/882be21e360740a5a3c8ca3912fdcea9 : SUCCESS in 4m 00s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/47c4defd9bae414293575a69760863db : SUCCESS in 12m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0743e88ced424b32b24cbd17e570e8c5 : SUCCESS in 10m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/01e529567d4348398bbffbd79d22f127 : SUCCESS in 11m 27s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/08b4985ed48b440bbe3c1b0a5ebc5fcb : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a9201bed9afe4e36b9740b0319dc2115 : SUCCESS in 10m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/86ce3764562f412ba84e0681e8661961 : SUCCESS in 9m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/60318076846845d8bc053c2d584fa591 : SUCCESS in 10m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9cbbf5b700b04097b7d2273503f2943b : SUCCESS in 28m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d89ea7294bf24926bf1216d7a4438f49 : SUCCESS in 10m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1fd7b3c81bbc462b9c98e38f09b5b3e5 : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9ecd349dc0594d6a807899591683bebe : SUCCESS in 29m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8973342dc28a4f758a0c18881a0d2be5 : FAILURE in 12m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e308b5326de4284982fb02f78878e31 : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2a24b2de71e349cf95f069ecc18855e2 : SUCCESS in 27m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ced96ade17f7404794bef2b1555709f4 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b51ce16ce99b4f5692ef7fc70c9da1b7 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/230cc0957da9439f82cc2699893d4d0f : SUCCESS in 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5d27406ef6a4b0ca97c5b65d35dabd3 : SUCCESS in 47m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/74cb91dd3a9247d48a1a1d6298231be8 : SUCCESS in 1h 49m 20s","accounts_in_message":[],"_revision_number":1},{"id":"ee37cb43f033a1c84362f880f8983b7503c8b1ce","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-02 23:01:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"74bc794af04bc14e337dddcedfd95c9e15afec25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 23:27:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/43419c782be043d8b712dc1b6c9cae4c : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5e31dc3f641395056ec6461d5548cdc64e1efa1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 00:28:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe9967ca72964bea9e636c4821c0e607 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9099b6e7a90b4c6c8fca92f787ce56e4 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d0bd3b6aa624176af8ea52d21a827d6 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd1401af034a4fd2898a6a98406fb670 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b836a76e54f74d23a00f60257c0eec91 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8ff46d05fbda4c16ba9350860616a41f : SUCCESS in 5m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7471972a135244788e8db543f8a41900 : SUCCESS in 4m 43s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/70b58aafe7e54353a00d428d29d2fa49 : SUCCESS in 11m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/94302ed52e1c4b7a891d716f72befb46 : SUCCESS in 12m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6093cc754f6e4a9e959b4cd704880ae6 : SUCCESS in 11m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7c958dcaf8ff4cfd86fa044ed8455404 : SUCCESS in 6m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/11038adae72c42dfb7845c9587e92877 : SUCCESS in 11m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7039717f28a64bfe8caa895b8b725720 : SUCCESS in 11m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6828e1d9b11e4517a7c581d415b78c2d : SUCCESS in 11m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7feffbdd0f2d4600b299e32a44d7cf61 : SUCCESS in 16m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/557a5f89847d49898b9e25d39acc8f2c : SUCCESS in 11m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0f930ba81ca84bd6b64bde316dd45843 : SUCCESS in 6m 18s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/33ed0532d7674827a89daa820ff64505 : SUCCESS in 22m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ffd0c6a32dea4529b916458df2247be6 : SUCCESS in 14m 31s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/64d48cc775ae42bd9800d2c28e74cb28 : SUCCESS in 5m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/286e9a0b528d434eb65648f0f9adb861 : SUCCESS in 19m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e5c65ed74bfe427e84cb43eb1891bfee : SUCCESS in 6m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38ded99b8a7e49e7811ea9101eaadf46 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ab4f6de6f5f34e70b6aa45802e209a89 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87378e18b1e34e288b26f4a9b0ed3019 : SUCCESS in 46m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07565307df444898944f532439221d2e : SUCCESS in 1h 25m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b7c439bac9d06798a3ced5c67b413168a4f47e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 00:29:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f9bb16dfd04da26da7f5e705c01bf7455c4a910","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 01:50:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29c27da8002434fba048e46171bb368 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec307c602f164bf8aad153b820a987f2 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e6d1b1ff804bbba8b675419b212e94 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fb92694c20b4800b5057ee7534359b0 : SUCCESS in 5m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b17dcb45785e41cfa4504ab00d7caefa : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0d6285a0f8d0439aab0f478c6a2cafea : SUCCESS in 4m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a21dd95f4109435fb6e0f4ea625ad2b7 : SUCCESS in 10m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8b1f8199e19549c9b3161a12fc0e7274 : SUCCESS in 9m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3c80e144962745a297789fead25b8fa4 : SUCCESS in 10m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/850366cebb184ca3b5a029b3604d4c3a : SUCCESS in 6m 12s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3c4ba36fe4394434b6b81d24275d2868 : SUCCESS in 8m 33s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0dfa9e428d0d4f7089e06527b7d39239 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b0407c1ea90d4db39550dc8e763cfb0e : SUCCESS in 10m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bbba717758f34c82b17af7d8127be457 : SUCCESS in 14m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/62b8da8499ce42cda89e6105bf06d0f7 : SUCCESS in 10m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c769d0a895be45468f866f8e6ce6b72b : SUCCESS in 5m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d4b707b0ba28476480dcb704b848e50a : SUCCESS in 23m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/780dc65ea3d441fd9836b6da1b8badbe : SUCCESS in 12m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ff02fc44cd7146d0972ac46ac2192585 : SUCCESS in 4m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0aecf9586f649c5bfe49857a020688c : SUCCESS in 16m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4a35af0bbbd9408aad800ed5956dd7ed : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7795e5ec971d48eab5ec2bac8ea66791 : SUCCESS in 7m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/744db0b1780645139b78f78d702eec1b : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01ff493ea4244cedb86be402e0afdd27 : SUCCESS in 46m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9e9d059796a849388819a7edc6115c28 : SUCCESS in 1h 19m 23s","accounts_in_message":[],"_revision_number":1},{"id":"67203d1bb040c9fbb9162a5c9a0f33ea36dc001f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 01:50:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"162c2b642b8360e0eb00281b98ef33c11036f4e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 01:51:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7fab40af73b4b68a605ca0e41764d56 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dacf36952fd32bd29b14293aa8d2025f40b7c158","revisions":{"dacf36952fd32bd29b14293aa8d2025f40b7c158":{"kind":"REWORK","_number":1,"created":"2021-06-02 15:54:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/794260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/794260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/794260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/794260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/794260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/794260/1"}}},"commit":{"parents":[{"commit":"154d6ee1381d3055970f0c4866e42520a21efc31","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/154d6ee1381d3055970f0c4866e42520a21efc31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-02 15:54:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-02 15:54:05.000000000","tz":0},"subject":"update constraint for openstackdocstheme to new release 2.3.0","message":"update constraint for openstackdocstheme to new release 2.3.0\n\nmeta: version: 2.3.0\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Commit: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Change-Id: Ib616837841a5d470b4d3fb1ef710fea4c8ead4c3\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I728024375b4d4d951251872a8e6b5977f59e710a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dacf36952fd32bd29b14293aa8d2025f40b7c158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dacf36952fd32bd29b14293aa8d2025f40b7c158"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
