)]}'
{"id":"openstack%2Frequirements~794344","triplet_id":"openstack%2Frequirements~master~Ib9132924929f9f04921c5137d8cfc08547892f49","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ib9132924929f9f04921c5137d8cfc08547892f49","subject":"update constraint for python-manilaclient to new release 2.7.0","status":"MERGED","created":"2021-06-02 21:55:50.000000000","updated":"2021-06-03 05:42:25.000000000","submitted":"2021-06-03 05:41:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794344-new-release","meta_rev_id":"d4c7c4793d0aa8a314f4772b697efcb58b7a3de2","_number":794344,"virtual_id_number":794344,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 05:41:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-03 00:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-03 00:56:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 22:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-02 23:24:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 00:56:56.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a65d962fb534b2a66fad496c3febd467af6016d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-02 21:55:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e171c7088a57929aea3f0f4442d8559d31ad889e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 22:12:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7ca57d5e43034cbb99b553ca9720ebb2 : FAILURE in 14m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bc3f5bb953c75324c4be9782e364a0f019d001a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 23:24:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66045a3133334a11a8081a59161c5aa3 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5466c2854b744f57b617f6a8071c6a90 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a651d21642b34a70b3b6c7335cbf172d : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c37f0e56fd3c4e7f8ac5320c7772b3bf : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe0df57e538f4c669a62011f41e00f10 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0486e1a9626f40d18cb2d0b2acacabc5 : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c5894363d264d4084fed7912fd7881f : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/da29987fe7dd43aa80cd02b363dcf493 : SUCCESS in 10m 05s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d2fb9c647da41c480f7d758ceb8fde3 : SUCCESS in 10m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9f1819b5d99643ea923349281b3a8c32 : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/458b687882af4e5cb1fadc6a4b47db81 : SUCCESS in 6m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e25085327e484e648a9e497957571879 : SUCCESS in 8m 35s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1b31ac63cd8e40109ed1c1c92b6a1037 : SUCCESS in 9m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/61dcc64c7a7b4d28ac4fcd63e3084793 : SUCCESS in 10m 44s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cdee5b0caac04f27a8838fef610d39fa : SUCCESS in 14m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9513f0767dea4674952d2a40a2b05086 : SUCCESS in 9m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dd67b19490e44d7ca5e53039b05fe4a6 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/95da622d0ca040d692052e6079ab5a58 : SUCCESS in 27m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/23f43fb2a19b447d946ada262ec4f61e : SUCCESS in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/92c48c796e25484b9629889553ada90e : SUCCESS in 4m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dde3ad9db53c45b18609491b1d092ea6 : SUCCESS in 17m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/043a50e28c04419bba94461f3ef387c0 : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f7818132d8a48b49821e3dc578b93f1 : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/08a1a7196f83467896423e93f1da016a : SUCCESS in 20m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dc3d7f633b44063b648df3f060a4ffb : SUCCESS in 46m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3507bb85d57f4ce5b0baf487ce0b781a : SUCCESS in 1h 26m 37s","accounts_in_message":[],"_revision_number":1},{"id":"844b05c2607d623c8cf0f8fbd5d2f8a266f1b895","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-03 00:56:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bbb4916001c17ef10aac0652106d0b3d13518667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 00:57:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ab49cb5e1569fce6d92a110a85e2e8327a99e37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 02:30:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63171a752d3d4479a97bc3e467271f0a : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909b17d91a024a069f6fcc07bc4d3728 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdcc21ef884d49a6943a5b4c7a14b3cf : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72fb0e467d424973816822496347fa94 : SUCCESS in 7m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/097d43eadfa44590afcff5d390c5a38a : SUCCESS in 4m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c703ee01b8fe48df9802e3ef214dd020 : SUCCESS in 4m 45s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/37316f8f75414d13aa48721b4c1707b5 : SUCCESS in 13m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/149a6f297c1043c3890821e41b46b1b3 : SUCCESS in 11m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/51631f49f8714621b8f526b7339f781d : SUCCESS in 11m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c1e24bed579c4d2aac5ad13c6ecaeb9f : SUCCESS in 6m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0846857f7cd5428f8ce8129d9d4766ec : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eec0b7f1542e4413af09b3095125fc77 : SUCCESS in 12m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0acd89a357b7401da7e83630d9210678 : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7fc0ef4be8ad43a88db4ae951dc190ef : SUCCESS in 16m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2a2ce2c26a2a40d6a454fae3fdbb5757 : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5b57d20cdd8c4273bb0e9d07d1b46c3d : SUCCESS in 6m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7f36bc5244ba41529b4e7385ebf0019e : SUCCESS in 24m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9576949a127c47f79620e84e2aad7f9f : SUCCESS in 12m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/47cc127154834d43ae39f7e3cbfe16cf : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e60a3dc9f07b425ab78462665ceee0f8 : SUCCESS in 16m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3234c07bb61646d18ecd33de9659a6ce : SUCCESS in 6m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/abfc5e502e5548f0b71d301a196f5acb : SUCCESS in 7m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5c61b2eb3d9d4a4d9e6eb6f3d6a90833 : SUCCESS in 9m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da476856800e4828ad02703ce5d7c0ca : SUCCESS in 46m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51842c2417f74ef8beb6fc5420530549 : FAILURE in 14m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f26dfeef812b5dad035c2d5314e6d158853ef4b9","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-03 02:50:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"871fd6e1e1ac7da11c393c5b6107aca99b53b060","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 03:07:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/96140a50f89b4a9b969fbc910fbac51e : FAILURE in 15m 43s","accounts_in_message":[],"_revision_number":1},{"id":"daa6b983be5dc573dd80680d90053a84121100d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 04:23:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa1e30ed6e042c8845651b8272a98db : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadfbe8e711c4b2285704b5fb9868f2b : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67268ff01b6f4be6929554a62dec0654 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3353770c624a477ea473f4029320ccae : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5cd394b1c3440029fdf4f087b81a222 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fe1884ab10e34e8591a54e5449aeab65 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ae22a69cc0314b88a42f828eeb45ca76 : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6cf8a47a65ef496d87fe6c6a70cba422 : SUCCESS in 14m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/26738c419b784f2fa0f072bb61016ccf : SUCCESS in 14m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/da3417eed0964e35bd1a25bd9818b0fd : SUCCESS in 10m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/89e1080325b24347b42fd2d40b5ef46e : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ca071cf6c24a49899619c481bb797b7d : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/73689660a5904ac690f179b6a2b45762 : SUCCESS in 9m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7034dbf54b6f4f69b95703d152630b14 : SUCCESS in 11m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/55fdb591961f41f68188410d94b84b7d : SUCCESS in 18m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/731115562c24401cb324d77dae252875 : SUCCESS in 10m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5a58b0008e349fc86371708d9c0c921 : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6517f24c0b2945dba3ba77dab4100847 : SUCCESS in 27m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3219086e8ff64a3aac2f7c2cf0fe64b5 : SUCCESS in 17m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ac5544c1e15144399c06a017ec2bb231 : SUCCESS in 4m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45a8bf8fc19e4933b001165ed9db28b8 : SUCCESS in 19m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a7d934a9c77b41bd953c1a713eb86a6e : SUCCESS in 7m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/95a6731d2674406091fa0ebb7200255c : SUCCESS in 8m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9d0b03d4fe9b49299fa85620f11011c3 : SUCCESS in 15m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc66cd58b6b8462f88b9528c9abfd3a6 : SUCCESS in 45m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7147949a5d4d4ea3b30da11fee9e9039 : SUCCESS in 1h 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"5f3793ba332605547096afed88c5d58a8d080b80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 04:23:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec06dc717bdf02d09b56e3b68de6a08b18a87aee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 05:41:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c097e80877e431ead599d39fb868bbd : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/183eb6d2c52d4a758a18fd980cc5049e : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdbe33a3c62a4742ac3776f82baaf5e6 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/154162097c9542c9b5a492e17ee50632 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1572fa51cdae4b3bb34cb748e7aa44ac : SUCCESS in 4m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/359c52e3b9164e5c8b6f0f019d8dd3de : SUCCESS in 3m 35s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/41b445f8d7b04330b8e4d3b737da8ce5 : SUCCESS in 12m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5bd353b24c794d2e932ca1eb66be367b : SUCCESS in 11m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/050934bf27244bca8195726b0aae3829 : SUCCESS in 9m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e1d26605270845fd9b78353d2dbadeb8 : SUCCESS in 6m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/147730b7d9224c9db7b618deef460de4 : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d495c319e5674f98b61151b18cadb605 : SUCCESS in 8m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a0416aa2bea849e28a8825e0c206adca : SUCCESS in 10m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a8f90ea417c04d69934ed71ebae69105 : SUCCESS in 12m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a147b0064dd940419829ea3ce99c676b : SUCCESS in 9m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5b64f6a9989c40d5abcb8f19aa879a07 : SUCCESS in 4m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/960127ea867541ac9283f8fa59b10b2d : SUCCESS in 24m 26s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/39556a170a584c36bee3956ae8cc4737 : SUCCESS in 20m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4338df6cf07c4b158b24b391069d2a68 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/064e45dc1f4641f2bf1fe4a223058f6f : SUCCESS in 17m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f11ee02739354c518f75e297ae6b65ed : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b6c9582c2bdf40afb81acd3da061acb6 : SUCCESS in 5m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bb41854df6764fdab92a9faa46bd3ec2 : SUCCESS in 10m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22e8624d44fa4694b2d3196f932ce165 : SUCCESS in 45m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c57b4c697b614eee8b3f2a0e05e3d880 : SUCCESS in 1h 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"b0ee44ba702853f9f089af55ed5419fe35830354","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 05:41:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d4c7c4793d0aa8a314f4772b697efcb58b7a3de2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 05:42:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b833b2699767480194011084fd1d860b : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1ae497746816131a86eaa068be9f96780115f9a","revisions":{"a1ae497746816131a86eaa068be9f96780115f9a":{"kind":"REWORK","_number":1,"created":"2021-06-02 21:55:50.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/44/794344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/44/794344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/794344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/794344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/44/794344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/44/794344/1"}}},"commit":{"parents":[{"commit":"154d6ee1381d3055970f0c4866e42520a21efc31","subject":"docs: Update Freenode to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/154d6ee1381d3055970f0c4866e42520a21efc31"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-02 21:55:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-02 21:55:48.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.7.0","message":"update constraint for python-manilaclient to new release 2.7.0\n\nmeta: version: 2.7.0\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Change-Id: I98005d186a4137417d3f131adffc8965c91e365b\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ib9132924929f9f04921c5137d8cfc08547892f49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1ae497746816131a86eaa068be9f96780115f9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a1ae497746816131a86eaa068be9f96780115f9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
