)]}'
{"id":"openstack%2Frequirements~794403","triplet_id":"openstack%2Frequirements~master~I652c88b4df194a123c0101109854f9c4d181332a","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","hashtags":[],"change_id":"I652c88b4df194a123c0101109854f9c4d181332a","subject":"Updated from generate-constraints","status":"MERGED","created":"2021-06-03 06:28:51.000000000","updated":"2021-06-04 17:19:46.000000000","submitted":"2021-06-04 17:16:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794403-openstack/requirements/constraints/noclob","meta_rev_id":"c637beeb1a1a5e4c876af80249667ab4bb092886","_number":794403,"virtual_id_number":794403,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-04 17:15:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-04 14:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-04 14:50:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 06:50:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 08:30:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 21:38:47.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"f338107fb4aca6adb88873f0fb5dff255d286124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-03 06:28:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7628c89b5a5c4f0f4e9aaf54a7689740938731bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 06:50:28.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d8fc487def574706971f7cb778a93935 : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"5d52edb9d1170e6ca0fa56e9c5100b20193db2ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 08:30:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99630abaad424aab8393348c144f4a0e : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4928e51adfa94ede84ece26f8869bf68 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d5b2b2a14d846caaa3b553583c709c7 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53abbcf859414267a203ecb6f586059d : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/130fb32fa25e42b3ac055347bdca1180 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e27289bc2efe45e2be825031261941c7 : SUCCESS in 6m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78a27658cc724b00ad04d7ff6ff3cef7 : SUCCESS in 6m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/55235c374b854295b2014fed4ff771d6 : FAILURE in 7m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/741bef6c253b4e428b8dacd8e8c09585 : FAILURE in 7m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3a16bfeb5bc84d69bed0c2a015fb20c1 : SUCCESS in 11m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/16c83c3a63d3427db9de4b5bbe76686f : SUCCESS in 7m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ebf1d0274e424142b3456f38f9d86940 : SUCCESS in 11m 12s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0d249acfb7e04d0ca9145664f6784333 : FAILURE in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a59db415864b49c881859029891081ce : SUCCESS in 13m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/764907e1956f4ed2a3aabe9910a97267 : FAILURE in 15m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c360eed78b1243d58f06645d7a0ce160 : SUCCESS in 11m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f0b0b1f585894a6bb4d8131c798a1542 : SUCCESS in 8m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/75d5ecf0358d4e3684674e0572684c39 : FAILURE in 22m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/532e5e515dcd4170aa5c320d1111d1ec : SUCCESS in 15m 39s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f5ecce8353794764b254d64af0fbb2eb : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9c1c8cd4ec5d4d93a9eb0878282daf53 : SUCCESS in 20m 52s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bb384644674c49928c3b81b4dfc0ab1e : SUCCESS in 7m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fc768bc6f54469db5135d256da13d5b : SUCCESS in 7m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/24c3878108084703bfe756f403678abd : SUCCESS in 15m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d9d12f3c5a748f7a98b7c61c347f5a3 : SUCCESS in 46m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a898e63f266c420eac3a1dec729f1b07 : SUCCESS in 1h 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"404d3ec828ecfc8595041a09447a6ab2402528ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-03 21:38:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"59e79b3a9ca7bcd9f7aaece4c84f6e95ea41f7dc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-03 21:38:51.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"0698d1577609f5dc265fa3790ea9dc05523fe09f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 21:53:40.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/085410619e9d44099fd43525aa01d164 : FAILURE in 11m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3174b2cc38aabc52dcd2f1007fff42798b14d059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 23:00:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ced8cb08f864fd99aff31eeed268a3a : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1d0f5f83d904106a08c5c7ceaf8f3e9 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae5afe957f9e4bd2ab55d6f40290084e : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a00f4beb0d840b7bb64f544a77a861c : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63e5bd784eed4057b3facf39ae91daff : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fa80669fccd044e9a5d722e7ed7a0d02 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b4d18054e94148f499f9220d96f663f2 : SUCCESS in 4m 24s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/217ccbc705da40c69dc71cc8aee1ad66 : SUCCESS in 10m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd55caf0f4e642c1b82e53b3bba7d457 : SUCCESS in 10m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2583fee948af4deb8d96fadea1766db0 : SUCCESS in 10m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f46113767bc84065af47b6a03951ce65 : SUCCESS in 6m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cf5db89c341b40d8b9a3f38ff5ebf4ee : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/abc15c558cf841829fde08d31b0d3d9e : FAILURE in 7m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/0bb90bbb1fb7447789f4e6de80596bf8 : SUCCESS in 9m 52s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/22b978f1a3f742c5a3d94cc68ba76ae2 : SUCCESS in 17m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/222d82287783473ca68786272141ba59 : SUCCESS in 11m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cda52ea68c324e5ea9f8286dd943cf7a : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/dcb0e640f67a491bbb94238f48394e7c : SUCCESS in 28m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d1c6c9a66092498488b490abae0fce17 : SUCCESS in 13m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/510eb4e57b024cde8dce3026ea60fdf2 : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98ac2a8e78ec478f9ccf841313710ff2 : SUCCESS in 28m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dbc018e0e7da40058dc49efe4019e636 : SUCCESS in 5m 15s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f7e861be912146ed88684df8ef0996c0 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c9c51801b33c4dda810eff88399e9ae3 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1899be45be904c9b90ca60387a2a47ce : SUCCESS in 47m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e244059e767048878889dc71efe367d4 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":2},{"id":"eed907e7763357e1569359315e18a834804c2de2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-03 23:48:46.000000000","message":"Patch Set 2:\n\nrecheck\n\nprobably legit","accounts_in_message":[],"_revision_number":2},{"id":"4749c0b7d83ae51bccd4179a2e0a9b58aa054c0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 00:07:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c9c9ed54d4744b859eea0301a51aaf64 : FAILURE in 16m 49s","accounts_in_message":[],"_revision_number":2},{"id":"9e58ab2f438ebfdfe81f142cfddf50e006e1be94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 01:31:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962f22cca6f44cac96f4ff7c7f1b6fbd : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed2025e3dfc4890af81df530923f683 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e315fb5d68014b1b9e929f92b3a35f33 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e943d4ae47f4f308db645444a1b1531 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a43715b1f774b4ba2ff32d6adb02ed4 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/598c3e3b12924c4d909f63bbb1d98f41 : SUCCESS in 4m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f7e06e5414747c29c8693915b96943f : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/681fea21c04c4053b0524247a8f5d52e : SUCCESS in 10m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/db2131c136e94d21a33a2041481f5b4b : SUCCESS in 9m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/645dcc5035664480806b8d0fe5052e12 : SUCCESS in 10m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/670db990e85b4dbb8aa9eb79490237d9 : SUCCESS in 6m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/281597b7fabb40afa4953b2be87b766d : SUCCESS in 10m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/099579e56f0440a791495885ede49f14 : FAILURE in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4a82fdb9b3364a428c66cd3a3e242829 : SUCCESS in 10m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/afc4d28690514ab2b615814a63c4bb2b : SUCCESS in 19m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/45a12f41661948aeac001815447a7575 : SUCCESS in 11m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bdb7632e4ec2488f8ad253bdaa60a064 : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e2feb45abb414eeda4bc929dac68c9d2 : SUCCESS in 24m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aa170bc0b6c044ef91cf44306919b366 : SUCCESS in 12m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e62b6df4608747b2a12eff95d1d448de : SUCCESS in 6m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc2c0f125402468a9b8c1cedb99a3b9a : SUCCESS in 17m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/892c1805191a41da9cc24e4b5a2efe4b : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2c1a715153f34a0195a2887ef56c275d : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e7d973477d604d239a15102303c99954 : SUCCESS in 12m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74743fcb02634708a73fa3d305dcd6cf : SUCCESS in 48m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ebda7315bb474b2b8d6065cd32933343 : SUCCESS in 1h 37m 37s","accounts_in_message":[],"_revision_number":2},{"id":"437cfde90528d46936d2343ef8fc3f32f6a0751c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-04 07:16:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"89ab4869406f38f0b52d58411f88f9b9dbf2d608","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 07:31:40.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/40e17a74767d48d095a4b5f83b6507a9 : FAILURE in 11m 40s","accounts_in_message":[],"_revision_number":3},{"id":"ccaa97bc0bc3b3ffeeff1e097f1556d3a9ff0741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 09:17:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25141a74dc3540b7a4baad8327e560e3 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9b4ecbf37447d4aabba4c65504ac50 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5f702608d84d72bdebbfcde96d7f06 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dec4db0ef8bc44c9b7f298cd9e65f5a2 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a2e255628064b779eada7e72e1ebaf2 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01c52a0d82c341ee95003bcd7500159b : SUCCESS in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df0d1cc92f4d47ab8d51e321cddf2ef5 : SUCCESS in 4m 19s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/570846db2f2a49c5a6f0d5f0692aabd2 : SUCCESS in 10m 40s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/72212423c3c44aeb8c71091b1ba5d052 : SUCCESS in 11m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/183b7f0e2607433d812d554095e6a9a0 : SUCCESS in 10m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5f8742a467684a48a2ea17b37928454c : SUCCESS in 5m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4b7fbb78d56e434ba376a46b2257528d : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9b312bb87591414a9cc26e40f37adec3 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/77579210887243668267e20725df42ae : SUCCESS in 10m 39s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0e65aefd63a44a2b8c082d94181abe5a : SUCCESS in 13m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3860d17435a54c80aeb61a5df7b6d16b : SUCCESS in 9m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6c0b6e3b31fb407dbd73af54992aceff : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4ab59af109384a4c92929e1eebc8b59d : SUCCESS in 26m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8125d3c3b58e470399ed449b3e018b4f : SUCCESS in 13m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/17664f2e37e246c8abb0b72a4a7d73de : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/50f486836b9146c8b612671565dfb54d : SUCCESS in 18m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/af7a74ecdcf04cb096df415680fd6af1 : SUCCESS in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9f6fa20370be45e0987a2e4c9fa43e57 : SUCCESS in 8m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/84103ed8695e40f6b661b3d41c820e9c : SUCCESS in 12m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59a48d5ddd384b329192a5f6209fa19d : SUCCESS in 57m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adc1c76fe1714150972133a1dc95daaa : SUCCESS in 1h 55m 48s","accounts_in_message":[],"_revision_number":3},{"id":"2fa215b30fd7d206a5341116303c4f37d83e92a2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-04 14:50:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"df22e9f9367080bd0cde29766b2290e51b634c9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 14:51:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c585bf7255e3a8a0519c990110dfb811eb3943c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:15:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f7554692c64c1bbb68934b0439b6cf : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbb041f32674362834b489f141bcb68 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f93742f242fe456fb07ca5d28a6aac23 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47f9ca7302af4d169c29747c7352b1b0 : SUCCESS in 6m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9333be30cb3846b0ab6b94b73268fcf0 : SUCCESS in 6m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7f9e7d90c2f048039935e5e3c8eba011 : SUCCESS in 5m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f9f00dec3dd24a8ab36cb77d5bc4e229 : SUCCESS in 10m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/99436c29ee2b4e45a7b13620808c56a4 : SUCCESS in 11m 05s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bcf4c03aad7c47a4a7567ed7b3d645cc : SUCCESS in 11m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cc4a15c6563640bfb2415ab9bebf2df9 : SUCCESS in 6m 52s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4eb1d226c1f544e99c0dde10fee5d326 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9dfd0029fa944957a134e339907c46e7 : SUCCESS in 10m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/75ad5aad2fd64172a5a6b008fdb0c7c9 : SUCCESS in 14m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cf366120f8374a34b98be076c20ac91c : SUCCESS in 18m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c335780a81654397a968ee11060642f5 : SUCCESS in 11m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e9e4156a7dd64f6d9fe4e9b167ae9763 : SUCCESS in 8m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/28d63b6f58a5491eb9de71a92681aa72 : SUCCESS in 24m 31s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c7ec2018c05d4d9380662e1a466f308b : SUCCESS in 14m 28s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2113c4aecece4645bd82939631392281 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bc2b24a4ae254e60b799186ed5a48fa2 : SUCCESS in 18m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/63c12614b8694d66bb5c193ed503c800 : SUCCESS in 6m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e014bd7d47748d3b487fb81304da030 : SUCCESS in 10m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/51fe43018edd4b17882321796c8f7fc4 : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb825cfe71b547f98b7471e4c827c7da : SUCCESS in 47m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fd4535b4458c4cdda8ca7864636c46c0 : SUCCESS in 1h 36m 48s","accounts_in_message":[],"_revision_number":3},{"id":"98cece27317da300eaa4a8f757a4936af58dd4cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:16:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c637beeb1a1a5e4c876af80249667ab4bb092886","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:19:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb6dd157abc24829b773596c8c613c11 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"000af79860d04c84e708e72b23a997c526093967","revisions":{"acdbcccf37e81cfd5931e1ac334b04f7cb75528c":{"kind":"REWORK","_number":1,"created":"2021-06-03 06:28:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/794403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/794403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/794403/1"}}},"commit":{"parents":[{"commit":"55001be901bb174a9fdf6dcb0dcce110204f616d","subject":"Merge \"update constraint for python-manilaclient to new release 2.7.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55001be901bb174a9fdf6dcb0dcce110204f616d"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 06:28:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 06:28:48.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I652c88b4df194a123c0101109854f9c4d181332a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acdbcccf37e81cfd5931e1ac334b04f7cb75528c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acdbcccf37e81cfd5931e1ac334b04f7cb75528c"}]},"branch":"refs/heads/master"},"f9066c53a422272d11d88de31930af4df5e214be":{"kind":"REWORK","_number":2,"created":"2021-06-03 21:38:47.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/03/794403/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/794403/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/794403/2"}}},"commit":{"parents":[{"commit":"9ec728fc05f8f582043a6cb4f68e1dad50ed5b6e","subject":"Merge \"update pyroute2 to 0.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ec728fc05f8f582043a6cb4f68e1dad50ed5b6e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 06:28:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-06-03 21:38:34.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I652c88b4df194a123c0101109854f9c4d181332a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9066c53a422272d11d88de31930af4df5e214be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f9066c53a422272d11d88de31930af4df5e214be"}]},"branch":"refs/heads/master"},"000af79860d04c84e708e72b23a997c526093967":{"kind":"REWORK","_number":3,"created":"2021-06-04 07:16:36.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/03/794403/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/03/794403/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/03/794403/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/03/794403/3"}}},"commit":{"parents":[{"commit":"9ec728fc05f8f582043a6cb4f68e1dad50ed5b6e","subject":"Merge \"update pyroute2 to 0.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9ec728fc05f8f582043a6cb4f68e1dad50ed5b6e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 06:28:48.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2021-06-04 07:16:28.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I652c88b4df194a123c0101109854f9c4d181332a\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/000af79860d04c84e708e72b23a997c526093967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/000af79860d04c84e708e72b23a997c526093967"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
