)]}'
{"id":"openstack%2Frequirements~794557","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Ib18d07b25640315b3f46f9741e64e99216cb299f","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Ib18d07b25640315b3f46f9741e64e99216cb299f","subject":"update constraint for os-collect-config to new release 12.0.2","status":"MERGED","created":"2021-06-03 10:19:55.000000000","updated":"2021-06-03 14:01:44.000000000","submitted":"2021-06-03 14:01:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794557-new-release","meta_rev_id":"b82e6308bedc5762789686e4a08e2c36515bb9fa","_number":794557,"virtual_id_number":794557,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 14:01:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-03 12:19:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-03 11:33:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-06-03 12:19:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 10:42:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 11:33:37.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-03 11:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 12:19:51.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"ff673214cc1ea59a7fcea1aa0b6d5e3944602205","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-03 10:19:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"309226ec666303ad6dc2be88e18a28dfa934452e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 10:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d3e58fa6a8a94863a12645b334e9ca34 : SUCCESS in 20m 54s","accounts_in_message":[],"_revision_number":1},{"id":"591dc3ef8bfcf2cd580a202fb69690f188d309ba","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-03 11:33:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"43f1fff1b192ff08d079dc6ec0c35a86fd639813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:59:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09960eacea9047c383f6cd04fc62bf00 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca698e2ce6ea4e92b78e8e1a1bf374c0 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5496aa8e559c46508f64986f26daaa27 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a826d9df5c7345258e467dd3c06ab19e : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/97ac534722864149be99ed76e5221deb : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c2cfe0d557fc4e5ea03063b017258a70 : SUCCESS in 8m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/32218c54ba054b3baceaa515f3e98015 : SUCCESS in 8m 46s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/56bad77794ca4baa90696abda6479b93 : SUCCESS in 9m 16s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/66e800da337d429d916bb821c78c8f4e : SUCCESS in 5m 12s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/5b57f4b1e1734491bcaa62647f7fb1dd : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f055a48717ac4da1bccc3e8692ea983b : SUCCESS in 9m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/040dcafac2104fce8a5f3bb8685e4455 : SUCCESS in 9m 30s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a6a903ae199a4fc4900d5e96847071d1 : SUCCESS in 12m 07s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/fcf4c377ec6249b8b955a3d6d0f68c17 : SUCCESS in 22m 57s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8f65c6733f8847a0b2703faa976a589b : SUCCESS in 11m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/00074634a28443e2bd9c5fdca6b07e5d : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8631be2bf7024f038ad4cdd33d675a87 : SUCCESS in 22m 23s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/27535091396a492da2bee360dac0d6f2 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bb89aaad500b44e4b490d542e1f1d84f : SUCCESS in 6m 55s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/d7c15eec8b4a402ebfee291a6620e557 : SUCCESS in 11m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa950df1b85d47e19df39903cc02d4f8 : SUCCESS in 45m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/198bc26ab0204648b2c7db05288155ed : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"77ac853500947232f91fe7ff8ea900f58fd87912","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-03 12:19:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3123cdcaaa0c5e1aefaffd4be15aaa48e531099","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 12:20:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a4532af7934c085042301e183f5b862014242908","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:01:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9105c2d9c83743359707e5fa2e5ba006 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8acbbde5250d4d90b654aa12fad2dd7e : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33c2b94f63f4cc58ad9092c440de381 : SUCCESS in 6m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/664d66a40121416cb0254e64df1d2964 : SUCCESS in 5m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/606ebc7c073142cd956668af2d5f6df8 : SUCCESS in 6m 17s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6fb0293daa5f44e08ebfbc56eb72acba : SUCCESS in 9m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/66ad4fb6217c48898433bd6138832172 : SUCCESS in 10m 22s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/272755acebe64ebf9546570dd9ba42dd : SUCCESS in 10m 21s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/f5841bade2ba4491ba9e3cc09447fce3 : SUCCESS in 8m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f4b0b60b8bfb474587f2f221674a927f : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/83eaee621dbe498ab48007eb8acc273c : SUCCESS in 12m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6464cac5b01c412fa3a22b73e8e639bc : SUCCESS in 9m 39s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1e8251f7c32d44388a7df3ddff58f7ad : SUCCESS in 20m 22s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/3468f81233584dfd9f30e9dbfa2962a5 : SUCCESS in 25m 44s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/d10cb9b419dd4eccaf242fd70eea7ae1 : SUCCESS in 15m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/627ba97db5cd43e991367c6b7fde991c : SUCCESS in 6m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/99a86fed20924e01bafb62878cc9558b : SUCCESS in 18m 51s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/5829f5e8d8ed43139618a0e6ab02fac0 : SUCCESS in 7m 25s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc33633785a34d7992c008bfef0c9ab1 : SUCCESS in 9m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/4b625e402987401997207b205fbda5a4 : SUCCESS in 13m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb50f15b138a462c94401a68860a7346 : SUCCESS in 51m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1ac0be231c945f69cf19302e34463f6 : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b82e6308bedc5762789686e4a08e2c36515bb9fa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:01:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69e57895e0abe42fe05787d0f605f2814bb643e7","revisions":{"69e57895e0abe42fe05787d0f605f2814bb643e7":{"kind":"REWORK","_number":1,"created":"2021-06-03 10:19:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/794557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/794557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/794557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/794557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/794557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/794557/1"}}},"commit":{"parents":[{"commit":"87b2c6e81c58cd4a3bffdeee939461403fd78097","subject":"update constraint for python-magnumclient to new release 3.2.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87b2c6e81c58cd4a3bffdeee939461403fd78097"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 10:19:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 10:19:50.000000000","tz":0},"subject":"update constraint for os-collect-config to new release 12.0.2","message":"update constraint for os-collect-config to new release 12.0.2\n\nmeta: version: 12.0.2\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Id9cbbb4644a0ebd86e70eaa2bc0b8fec7ae51a1b\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Ib18d07b25640315b3f46f9741e64e99216cb299f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69e57895e0abe42fe05787d0f605f2814bb643e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/69e57895e0abe42fe05787d0f605f2814bb643e7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
