)]}'
{"id":"openstack%2Frequirements~794559","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~Iac622ba3b8cdfc3170cb6fe9262f427cd5deb07e","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"Iac622ba3b8cdfc3170cb6fe9262f427cd5deb07e","subject":"update constraint for os-refresh-config to new release 12.0.1","status":"MERGED","created":"2021-06-03 10:23:52.000000000","updated":"2021-06-03 14:01:37.000000000","submitted":"2021-06-03 14:01:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794559-new-release","meta_rev_id":"799333b09bf3b4869b84d6e0dd6299ebb36f5e22","_number":794559,"virtual_id_number":794559,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 14:01:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-03 12:19:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-03 11:33:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-06-03 12:19:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 10:51:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 11:33:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-03 11:56:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 12:19:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"a40e9be21cf8c17f4f6ff08f41893e469f9aa712","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-03 10:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a23aec1efee1e46a2d74a41ea0eb041f1be9a3ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 10:51:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6e2ac5480d964870a629f1cfbafd02d9 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":1},{"id":"35e4d0b526c1e68829f2300967f1e84e89c4cbc4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-03 11:33:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c7d319bb0c913bebda12e9f2301facb2ccd1f9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:56:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/069fb9b5373b401c83a3aae893450820 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c429d5d097a44be5844f099a124d0bbc : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fdcb137e8cd474e9df3bcb487f49187 : SUCCESS in 4m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4d50326c6d44134af133705fd672e44 : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4e9f8c4cf3eb432bbc584fc5d00d3662 : SUCCESS in 3m 53s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/39badeb61be24c56a861197f027aafb7 : SUCCESS in 12m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/70ccfdc5b18044bdab4eea3cf8e76b75 : SUCCESS in 8m 36s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e4e91ec59c3f4fb1ac1eb14124ef3718 : SUCCESS in 8m 59s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/6158974023af445ca8e9da62a32511c4 : SUCCESS in 5m 53s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f5f900c51d6c43fc9a9df039121d2e34 : SUCCESS in 10m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/49a7d110deb540b68f587a5c062f66a4 : SUCCESS in 9m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49d72993cfa44c8cb0888c8ab0aff56c : SUCCESS in 7m 05s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a87f80d12d1f4f40bcd634993c4fdcf3 : SUCCESS in 18m 09s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a3a0168c20a4440495f4826e7bc077d7 : SUCCESS in 27m 28s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/73e111dff16740e88b86ddca3fd5d17d : SUCCESS in 13m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/aa086bb790114e89bc7e3acd228db953 : SUCCESS in 4m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6534678e0a064f3aa3ecf4fca174b14a : SUCCESS in 19m 46s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/f64eb66897624f7188d3b5a40e138032 : SUCCESS in 5m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/75336cd6a35344a38f404dba9e8e2b8e : SUCCESS in 6m 44s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/e9863a123fdc484e900de271f5582cfc : SUCCESS in 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6f487ed16744c75899ad802ba308cd0 : SUCCESS in 57m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1c0d5744ce5468bad4e517112685c0d : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":1},{"id":"aad190a6ddb81099c72d39934c37038d9558e259","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-03 12:19:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"667e809f2be4a9facb87eb32284500dfa1f78ac9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 12:20:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa5481872d10040ef49a6b0e23ece2105b10f59f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:01:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be9d05173b94c0287a35c905b0864a3 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5612532d5a3c48a2821c0fb2fda9c91a : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7f58958412a4a9f9f898c5c666aa1dc : SUCCESS in 7m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d2ead8b03e194351a81224fdebca72e8 : SUCCESS in 6m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b829c558153c45dea1273a06315af0c5 : SUCCESS in 6m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf8fd711fe054382a1108d2646fff768 : SUCCESS in 10m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d980b61cb014c70a59ca86434847b3c : SUCCESS in 10m 07s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/4e1cf41195bb44858c4a76926865e546 : SUCCESS in 10m 26s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/43c4301c30b4402283d4bb4d32f1de9c : SUCCESS in 7m 23s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/46086ad641d2421295c2984dbf8d6cec : SUCCESS in 8m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8bc8910a54f24d10b42c0d33d3e7ba4e : SUCCESS in 11m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3350f71991c43eea8acb20a0880c937 : SUCCESS in 9m 43s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/49275f4dd2cb4458b66ae455e06bc72a : SUCCESS in 17m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/29734d43a7104950949b90f5a077e809 : SUCCESS in 24m 23s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/33813c2ca86243a4acf6b634fa890a26 : SUCCESS in 16m 23s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea1c8dd1d96e48a99664959b946fe8e3 : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5eab12c1c97842dab8e6d49394bcc1f0 : SUCCESS in 19m 03s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/e063f2fa06f14e76b424ba98d3a63b54 : SUCCESS in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/139659a96e03410ea449ed43ead68e5e : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/bc8a00bfdd5f41b1a729a11153e73f9e : SUCCESS in 10m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8d42243115a40f6865955c05557b8ef : SUCCESS in 45m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/881f3eff258b434897dd9a07ed5ea062 : SUCCESS in 1h 15m 37s","accounts_in_message":[],"_revision_number":1},{"id":"799333b09bf3b4869b84d6e0dd6299ebb36f5e22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:01:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0c429161ee78e9aba7031ed4748d34aba0da065","revisions":{"d0c429161ee78e9aba7031ed4748d34aba0da065":{"kind":"REWORK","_number":1,"created":"2021-06-03 10:23:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/59/794559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/794559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/794559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/794559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/794559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/794559/1"}}},"commit":{"parents":[{"commit":"87b2c6e81c58cd4a3bffdeee939461403fd78097","subject":"update constraint for python-magnumclient to new release 3.2.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87b2c6e81c58cd4a3bffdeee939461403fd78097"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 10:23:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 10:23:50.000000000","tz":0},"subject":"update constraint for os-refresh-config to new release 12.0.1","message":"update constraint for os-refresh-config to new release 12.0.1\n\nmeta: version: 12.0.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Id9cbbb4644a0ebd86e70eaa2bc0b8fec7ae51a1b\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: Iac622ba3b8cdfc3170cb6fe9262f427cd5deb07e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c429161ee78e9aba7031ed4748d34aba0da065"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d0c429161ee78e9aba7031ed4748d34aba0da065"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
