)]}'
{"id":"openstack%2Frequirements~794560","triplet_id":"openstack%2Frequirements~stable%2Fvictoria~I162bac86048115605b8ed19f8d6f31cd176b76b2","project":"openstack/requirements","branch":"stable/victoria","topic":"new-release","hashtags":[],"change_id":"I162bac86048115605b8ed19f8d6f31cd176b76b2","subject":"update constraint for tripleo-common to new release 13.2.1","status":"MERGED","created":"2021-06-03 10:25:07.000000000","updated":"2021-06-03 14:01:04.000000000","submitted":"2021-06-03 14:01:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794560-new-release","meta_rev_id":"8a2d6ae7b74c8e938616967369d7a6671d982df8","_number":794560,"virtual_id_number":794560,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-03 14:01:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-06-03 12:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-06-03 11:33:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2021-06-03 12:18:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 10:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-03 11:33:15.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-03 11:52:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-03 12:18:29.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"c6556ddc6cfb3fc362a1a3331db0541fb8e147a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-03 10:25:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aec9482f6c7f8ba863ac3627e2c84da188a7b4b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 10:51:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a6fe9ee33ae748598075892ea20f426e : SUCCESS in 21m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8cf0787e99290205b6598fac6de38e2cae7cbd41","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-03 11:33:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad3c092a97648f1d72944a3761a45f2cd29b3ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 11:52:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7021ab9ab0b466996ff8395154ffcf0 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/39270796f50440bab4f67476d979441e : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f8ff1231f5f440f9e0251d827034dc4 : SUCCESS in 6m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/519bfd09778e47f0b9c004aba5d919e9 : SUCCESS in 5m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/48ac355979c848fe8eed029e42ca801e : SUCCESS in 5m 56s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/653df284b4da45c7b06a333be2e64481 : SUCCESS in 12m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c86c841ecaf440e0ba7f9936b58f8ec3 : SUCCESS in 10m 47s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/73c7d2630a8246e8a62190e6d93e3b52 : SUCCESS in 10m 44s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/df883426d58644c0b37f2c1f0b6b7e1a : SUCCESS in 6m 39s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/98de8cbbcfb1480d9994093b57818062 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66f9b9ca135c4f409dfeb432f257f6ce : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cade4f8b7876407d8186ab900c16bb2c : SUCCESS in 8m 24s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/5c6de3d6ae9640d7ab52dd35cee6391d : SUCCESS in 14m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/ed2c2dc2b3ed4172aaa71ec9e5bae795 : SUCCESS in 21m 51s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/69527754809245138ddd925ec1b8ba75 : SUCCESS in 12m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b51b0ffd74cc4b5c93aa903955e9d9be : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2f6f9ddbe2ea40a394bd6900e6615be0 : SUCCESS in 20m 04s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/338e332948ed4d83878253abeada6364 : SUCCESS in 6m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bdd0c50ed2be4fe785b1d666a768cd16 : SUCCESS in 9m 19s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/323e897ab0e243858bc33169d2297002 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af75349c81d74a9cbd1ccb4a4d7b31ce : SUCCESS in 55m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4f4478b24254f1dac5c14db61c1ceba : SUCCESS in 1h 22m 49s","accounts_in_message":[],"_revision_number":1},{"id":"46e169523ff297978c9be397900cd097506fe491","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-06-03 12:18:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7287d261c42b539b8d6c8bb52cdbe909ec3f31a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 12:18:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2abbd073f38462ec82bc3c77d38cf30daa7a4aa3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:01:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969500ee5c9840ca929f2f61ff6da530 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbbd458b6cb4474291310c97d3abe0cc : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f0b3cfce9ed489a8e5e21b5542c2bc0 : SUCCESS in 5m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c61c3a17b064815bc15a52ef11aeaed : SUCCESS in 4m 41s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2ffd75c4df804110b7b68ea517d7d233 : SUCCESS in 5m 10s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/36f43faff21b4b3cb86b711a33b142e7 : SUCCESS in 12m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/83e2223eb8ed4d4380874ca1463525b1 : SUCCESS in 10m 25s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/e7f3ef80b81a46bab718c9da1291c104 : SUCCESS in 20m 03s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/d810ea5fc2cb4e3c897e655cbb2e379f : SUCCESS in 6m 02s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/da4f94acafe14a75be45f1fe8ffd1e7a : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/498c4b14fe694785b1cb67f902f62d8e : SUCCESS in 9m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dfc72377043f45789be4770a38c4fb04 : SUCCESS in 9m 12s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/bcd1470aae4f4330b199ae9ee7bd3796 : SUCCESS in 20m 54s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/a2705de6a4664e6e8da13d67c3702676 : SUCCESS in 22m 46s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/07c90c801c524f049fa6966b0333f3fa : SUCCESS in 13m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ea92adbb124642d092088b39886f8e10 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6ff3459c081e4945826f9342ba7157d8 : SUCCESS in 17m 19s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/981c757d83974ecdbeb3266954170426 : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4c7a6b82359845c4a8cbc884e5683a1e : SUCCESS in 8m 10s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/5eecc758e5fa4f9e898a140e056d2865 : SUCCESS in 14m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f2699b141e94e5e9460299e8ce12dd9 : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54722f9c1a3640878dd62be0b25a1de7 : SUCCESS in 1h 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"8a2d6ae7b74c8e938616967369d7a6671d982df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 14:01:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36ba017b84277d2268d4584b13a1a68c0791bfa5","revisions":{"36ba017b84277d2268d4584b13a1a68c0791bfa5":{"kind":"REWORK","_number":1,"created":"2021-06-03 10:25:07.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/794560/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/794560/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/794560/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/794560/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/794560/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/794560/1"}}},"commit":{"parents":[{"commit":"87b2c6e81c58cd4a3bffdeee939461403fd78097","subject":"update constraint for python-magnumclient to new release 3.2.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87b2c6e81c58cd4a3bffdeee939461403fd78097"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 10:25:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-03 10:25:02.000000000","tz":0},"subject":"update constraint for tripleo-common to new release 13.2.1","message":"update constraint for tripleo-common to new release 13.2.1\n\nmeta: version: 13.2.1\nmeta: diff-start: -\nmeta: series: victoria\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Commit: Marios Andreou \u003cmarios@redhat.com\u003e\nmeta: release:Change-Id: Id9cbbb4644a0ebd86e70eaa2bc0b8fec7ae51a1b\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I162bac86048115605b8ed19f8d6f31cd176b76b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36ba017b84277d2268d4584b13a1a68c0791bfa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/36ba017b84277d2268d4584b13a1a68c0791bfa5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
