)]}'
{"id":"openstack%2Frequirements~795097","triplet_id":"openstack%2Frequirements~stable%2Ftrain~I3be40702110255f181351f08537284234d9f0673","project":"openstack/requirements","branch":"stable/train","topic":"new-release","hashtags":[],"change_id":"I3be40702110255f181351f08537284234d9f0673","subject":"update constraint for keystoneauth1 to new release 3.17.4","status":"MERGED","created":"2021-06-07 12:42:51.000000000","updated":"2021-06-07 22:14:13.000000000","submitted":"2021-06-07 22:14:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795097-new-release","meta_rev_id":"f05eff0d139da2833dd90b92a08fa3bbb4a811c7","_number":795097,"virtual_id_number":795097,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-07 22:14:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-07 20:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2021-06-07 12:50:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-07 20:26:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 12:50:03.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-06-07 14:46:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 20:26:16.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"34fc5520c97ecf0a227d9144e12fa4a439f4e319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2021-06-07 12:42:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61cd787507df6bf2c40d7c474153fa9041ffe8a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2021-06-07 12:50:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b13faa4f7f61b627c15473f0cf85b9e4f50808e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 14:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e3845c05b841bfaed4096b1cc0b65e : SUCCESS in 5m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3204c401c11343fd9a3b2b459352a54f : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67e1b7ce7769499c85c9924765e996c7 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4e0a1cebf724b9688250adc2b17decf : SUCCESS in 5m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/47c03f53c5374d69887475762cee6500 : SUCCESS in 4m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/42974c4c19764cbabff28c7efd7bf409 : SUCCESS in 4m 35s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/67b9b0c7bfbc4575825710bc6c8b5689 : SUCCESS in 8m 16s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/cbdedc3eecf442f0ad93bcc006eb98ca : SUCCESS in 11m 47s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/4e20029f3c98470dac18dd6a2d6bb07e : SUCCESS in 12m 44s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/9025a76d72dc4df2b9f3ac362db5f82b : SUCCESS in 10m 54s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/26282d1c084e4be0ac546abe5d0c3b7a : SUCCESS in 7m 55s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/58f1445193cc4302a42f25a26a6d0a4a : SUCCESS in 6m 32s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/486ade6d160a4a1da1b30138d771d6cd : SUCCESS in 10m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/0bd99c93fcf4482897a933acf0069c61 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4fb757c4fb804c5081113c9dc2f2213d : SUCCESS in 10m 00s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/0f422c9ed066418bb62d0f2c33c81d60 : SUCCESS in 14m 14s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/c83c760978f9434a9671bbaa3e32fe38 : SUCCESS in 15m 53s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/6b9d64cc9fc04e5e987211d9efee6685 : SUCCESS in 29m 27s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/cb80e31dc2ca493cb837f4ea9aca5b05 : SUCCESS in 20m 55s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/ae3c82a4d316499c8b7a8c500b9bbef9 : SUCCESS in 14m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/7901f07ee3264741b849fc5033814d44 : SUCCESS in 11m 59s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9745c550ed4648438ca7a54b019a7d61 : SUCCESS in 15m 07s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9b26f58e642440ea8946677005f87eb9 : SUCCESS in 27m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/71ccc3109a314187b47e0b37ec8641e1 : SUCCESS in 6m 29s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/bcfc79bf579f48c8a1549fb3ba32fe62 : SUCCESS in 14m 00s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/7e22828340dc4259bc744b1c0263a1ed : SUCCESS in 13m 39s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/d34746893d114b0b9e811061a1062f1e : SUCCESS in 58m 28s\n- tempest-full https://zuul.opendev.org/t/openstack/build/981b07d938214e59bf69d1b7b62e5224 : SUCCESS in 1h 30m 41s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/7a0d95477eff495d904adc94505e624a : SUCCESS in 46m 07s","accounts_in_message":[],"_revision_number":1},{"id":"84a1d5e7bcba1166f08ed7dbc4abc13c1d9dcc30","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-07 20:26:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb3b2fed53811c1350cb456c73db7da8b69b40d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 20:26:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28f2be44421c4a2f0ca00933a39966cc7e0431b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 22:14:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e60ff942c7d43e7ad2bac0e34ed79c9 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92c9fe9a49124b51af6400dd41a0d735 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1e76ad0193d4b3abdbcef05c7647c87 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e20425502b23417d97804bfcee8b6d76 : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6b154b36ecb54ac0aa268ed9f792f148 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9c71dc19ad44b7da8bf23cadb5e0631 : SUCCESS in 4m 04s\n- requirements-tox-py27-check-uc https://zuul.opendev.org/t/openstack/build/7f99ae99132d479f9fb6c2fc71e94dd4 : SUCCESS in 7m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5a312359b34f4085a770c4263f83569d : SUCCESS in 9m 00s\n- cross-cinder-py27 https://zuul.opendev.org/t/openstack/build/e5f4ce28d3cd4050815000e82cf36913 : SUCCESS in 11m 15s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/8da5856be5db4970851678369105b86f : SUCCESS in 9m 41s\n- cross-glance-py27 https://zuul.opendev.org/t/openstack/build/183083a32d6a416e87a3b98345e165fb : SUCCESS in 5m 32s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/4231a4c05f004b7190b48c4e3fb04b52 : SUCCESS in 5m 25s\n- cross-horizon-py27 https://zuul.opendev.org/t/openstack/build/3f2dc1f4bf7c49c29c0b2612a34fb839 : SUCCESS in 11m 36s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/f9045d282a704bea9f3079d5ab1246d6 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/27dcf8ef7d7c45c4ac1e78bc1a3ad0a3 : SUCCESS in 9m 41s\n- cross-keystone-py27 https://zuul.opendev.org/t/openstack/build/8a2b84b083ff4fb59b329ae2e5d2aadd : SUCCESS in 20m 36s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/a9905f7edd8742a2bdd9aa2999ee1cfb : SUCCESS in 19m 37s\n- cross-neutron-py27 https://zuul.opendev.org/t/openstack/build/b1c0e5ee8ffd48a7854c3613fa98a538 : SUCCESS in 23m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/454494970bbc423e8443e2bec69bd5ab : SUCCESS in 18m 22s\n- cross-nova-py27 https://zuul.opendev.org/t/openstack/build/d69963ef625640fd8c1cc4c620ff39d4 : SUCCESS in 17m 34s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/eabfc9a389ae43728ef73a5702890fe0 : SUCCESS in 14m 44s\n- cross-nova-py37 https://zuul.opendev.org/t/openstack/build/9eb5a5abd86d45b3bdcd16c099b78f18 : SUCCESS in 13m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/33d3ce8a4c5244229d2339fb4f8178e7 : SUCCESS in 22m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/524a40fcd4d84bd7a1c9e55e2a3bf079 : SUCCESS in 6m 44s\n- cross-swift-py27 https://zuul.opendev.org/t/openstack/build/ca57920ee4c54ba0a77daf08c7620a45 : SUCCESS in 10m 48s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/dae88be2beab4a32a70aabd8cda41ed9 : SUCCESS in 12m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/3e24db0ad8b34a1383e998e950f7f916 : SUCCESS in 57m 30s\n- tempest-full https://zuul.opendev.org/t/openstack/build/c958b7e610734ee3ab584c4c7c2def83 : SUCCESS in 1h 28m 07s\n- neutron-grenade https://zuul.opendev.org/t/openstack/build/a8c4230fb06d4b92bbdd84b8ab5f0777 : SUCCESS in 1h 10m 08s","accounts_in_message":[],"_revision_number":1},{"id":"f05eff0d139da2833dd90b92a08fa3bbb4a811c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 22:14:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9874f188560ba932c7fadc185caf8da3ed75b17f","revisions":{"9874f188560ba932c7fadc185caf8da3ed75b17f":{"kind":"REWORK","_number":1,"created":"2021-06-07 12:42:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/795097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/795097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/795097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/795097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/795097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/795097/1"}}},"commit":{"parents":[{"commit":"fd3ac38e886bc0f74ac4c665585a1e6291a1aa54","subject":"Merge \"update constraint for paunch to new release 5.5.1\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd3ac38e886bc0f74ac4c665585a1e6291a1aa54"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-07 12:42:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2021-06-07 12:42:49.000000000","tz":0},"subject":"update constraint for keystoneauth1 to new release 3.17.4","message":"update constraint for keystoneauth1 to new release 3.17.4\n\nmeta: version: 3.17.4\nmeta: diff-start: -\nmeta: series: train\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta: release:Commit: Kristi Nikolla \u003ckristi@nikolla.me\u003e\nmeta: release:Change-Id: I7d1082a517e7b74c28ece8a54c7cc42268144321\nmeta: release:Code-Review+1: Kristi Nikolla \u003cknikolla@bu.edu\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I3be40702110255f181351f08537284234d9f0673\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9874f188560ba932c7fadc185caf8da3ed75b17f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9874f188560ba932c7fadc185caf8da3ed75b17f"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
