)]}'
{"id":"openstack%2Frequirements~795270","triplet_id":"openstack%2Frequirements~master~I208b00629493953a502f8b95756820ae0d9a9cb6","project":"openstack/requirements","branch":"master","topic":"add_ruamel_yaml","hashtags":[],"change_id":"I208b00629493953a502f8b95756820ae0d9a9cb6","subject":"Add ruamel.yaml to global-requirements.txt","status":"MERGED","created":"2021-06-08 10:17:56.000000000","updated":"2021-06-10 18:15:43.000000000","submitted":"2021-06-10 18:14:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795270-add_ruamel_yaml","meta_rev_id":"9a62bd7b962bbf2069283067e016f52a8c5c4d15","_number":795270,"virtual_id_number":795270,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 18:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2021-06-08 20:38:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2021-06-09 16:52:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":1,"date":"2021-06-10 07:58:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2021-06-10 15:57:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-08 10:20:56.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-06-08 10:20:56.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-08 10:20:56.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-06-08 10:20:56.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-06-08 10:20:56.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-06-08 10:20:56.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2021-06-08 10:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-08 11:18:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd504289d191eceae1b1049e8d0d1e221210fbad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-08 10:17:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"81548059833f51ea2ba8dedcfc3df5ab671d892e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 10:36:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/aa2bbd5f2f8947ceb83eb11a2c62a681 : FAILURE in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"1ad9fe323e46cc703ac90c273fa79d9d71fe9998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 11:18:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/363a94e351094c4dae4504ace537f2fb : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea94ce73a28940548cf853834e13597f : FAILURE in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4fa539f344541d2af20234ac86a9f28 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bfd1c1916e6474eba3309274884d37a : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9681ca460ed24e5ab20d86cd9f4c5182 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d9fbec24c8454e33a5fd1d07e4633814 : FAILURE in 6m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d3afe771302e43b9b7455a73180e0c19 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/feaf7e9e46d04007a7ca42f5e435e641 : SUCCESS in 10m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b37ae4525ff94216b855ea922014691d : FAILURE in 5m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9c22c61ef86a486d97d3ad4c0374aaff : FAILURE in 4m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/634052d639354e0987bfaa4043a03045 : FAILURE in 5m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9fc62616c2a8427c936a985c7e3ab159 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8c6022c7df904224a242d111191bc8f1 : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9fb93433af24418196809700e9c08da8 : FAILURE in 4m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/bf189597f77b400c9c967aebf80c9e19 : FAILURE in 5m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f55bb91554d44683960a155fa19c406c : FAILURE in 5m 03s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c25b166cf43c427ea08a713aaad71696 : SUCCESS in 6m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/991367612dad4565bb0c80bef2acac1b : SUCCESS in 28m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8a471fd44f334f18904ac4b8f149e8a4 : FAILURE in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c0004c434cc942b5b9e799000b51d187 : SUCCESS in 4m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3e06d37e75e4ea2b974b8c6b36af13b : SUCCESS in 19m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/066678d5539d4ffb81dac1aa44da5af1 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e963b030774947cdb6a0c5cdcc9dcbb6 : FAILURE in 4m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3c5b081660284ec18a3db42926f8fe89 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f560d0059f294678ad535e5bf7e2acbb : SUCCESS in 55m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f2f4c73659dc448285b6f88e70577833 : FAILURE in 13m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ac00977a4fde17faa907e79777499407f563b0f9","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-08 14:54:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"18d6e39baa0c7cb12a7f75acd95d0e0b4304ea8a","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-08 14:58:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0690ad628cd3d403707b193fb27ab8ad5afbd47f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 15:15:01.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b485e9db513a4bf3838ac74bbebcd63b : FAILURE in 15m 34s","accounts_in_message":[],"_revision_number":1},{"id":"1ef2d7c2fc5bb105429be3e6fcfc7045463c78e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:28:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d984d35cf1b344c9abff06f13884c575 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5acc8d276e4c88a40f9c1e1e46fd16 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa12531482cc49faaf0a475973e097b8 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcb969a0ee4c48eeb9ec4dc41db0430a : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8896f178d514a978d8c84c3161ea51b : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85db1ede644947f09b95dcbe61cd6364 : FAILURE in 6m 21s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2564f680ee3a4a0aacdc1be13fa997cb : SUCCESS in 4m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c45f400afc2f41f1bdd855ecba178d22 : SUCCESS in 12m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7e0091d2a8db4c9391ef49a4cfd9fe49 : SUCCESS in 11m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/132f1fe5d7a54c528be69fed44446e66 : SUCCESS in 10m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4e9c6ea16b854ba5bc3ccc4d13103f83 : SUCCESS in 6m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4ba5df15f33846a38a501fe9d1605445 : SUCCESS in 12m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/06d0631ff47b4470a6ddad1e56882832 : SUCCESS in 9m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/443a878013714157a2e1b189763fa198 : SUCCESS in 12m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aa70ce7750304ee0ac37aff0c73a44f7 : SUCCESS in 16m 03s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b860c64f79ff472382fc4a761a92f6a7 : SUCCESS in 8m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dd3483b386bd4eb0a261dfc0cfe47134 : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4660d95410ff44ada9100cf905325f2f : SUCCESS in 30m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0d8fb849ee264bcf87f535f9a747dbd2 : SUCCESS in 13m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1de85bcd1c30423daafb41cc0407143c : SUCCESS in 6m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/02022982ba4049c689071169e7639e91 : SUCCESS in 20m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8c509e11a041460395bc0467b6527c5b : SUCCESS in 7m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec954db54e1840f79e64b250ed17627a : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/91b38b6509754882b89ca17cbe3773c9 : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c958105a05d042de92b397e86369477e : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a3203eecbd04b9ca09f82287b57ff32 : SUCCESS in 1h 36m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a6032dcecbce66546098663f66e334eadba83d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-08 17:38:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fd353065c8ef03ca04f61945cabf00c0f1919f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:56:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4eac1079a11c43bb80be3cb634c95a2f : FAILURE in 15m 57s","accounts_in_message":[],"_revision_number":2},{"id":"3accb21acd8caefbd2a339414cd9ff2ed10edc89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 19:17:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32b2a388c79246fa8ee34fb58a5b6062 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c985215f9a15416681ad37ab7e8a6b46 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81f6ed70a0194adcbccf6a066b1e8fc7 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d7e02f8c32f41339215cc3f90d118e4 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/879610cfd9ba42598b6c5d4f2b3e9c9b : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e7dda40b79b483e86136b756a1dba9f : SUCCESS in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5df47bb4d07e46699e28ffda274406f7 : SUCCESS in 4m 13s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3c7b5f569a2e4535805680c913dfda3e : SUCCESS in 11m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bbc21b0a8004493f958c5d0c2c83a5e3 : SUCCESS in 11m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/59ddfa85b76d4330b9042baf55bfe331 : SUCCESS in 11m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ac4e40b5f2ca45c2bb92714350c9a7e9 : SUCCESS in 6m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8757c0feb18945e5bd4fab41b37150c4 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/920e69f6ab7b4abdae64974556c170d2 : SUCCESS in 10m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c127e39442f4425bc3de9352aa13947 : SUCCESS in 10m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c7c8664591b64bed8f0574135bc332ce : SUCCESS in 13m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a3e06812ca56439ea9c58ac4b56ec760 : SUCCESS in 10m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cab8407609ac48ef9786fcc7f24a07b8 : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d0e0cf966c844326bbc0ae0624edc6a7 : SUCCESS in 28m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ecab80e8997b43ceb0305e9b28eedb3e : SUCCESS in 12m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/338d2c76c3ca410ea6f0708cbdebb3b3 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9b187123e4c4fefabbe618ff8a0b4eb : SUCCESS in 19m 47s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3da5b22ad8f2436b83ca00f081caf9f8 : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8966e7c759064a6facc71b3d4d964642 : SUCCESS in 6m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8a2a8b573b434810a2106109382f5e4f : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3558b40e7b934ed0b978b591b2a0611b : SUCCESS in 49m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8dfaa5ba688c41768283d28fa60c3d4a : SUCCESS in 1h 32m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2d67703074d9f575a765a3d05f35ec18832cce46","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-08 20:38:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"21927b65a4cb1fdc5fee9ee8ea9a4bfabe1f79f5","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-06-09 16:52:40.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b26045809dced24b13c6cab3bd861da0f62844c0","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-06-10 07:58:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"de5bfc7665dce2b67ca3653b000e984a9c459e98","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2021-06-10 15:57:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b4c27881746615f68b57a64b963f75f97fb5e411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:58:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7055468b062e00d2329c3697ddc934d75fc355e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:14:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7347c69fdc4d018b14546826c24334 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c5c626f6a634a988757c3bc7b0526ba : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e89c8f79b1eb4c7cb2f30e9b8c494a65 : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35b2b7c06d64440d9ef795d93ff15d30 : SUCCESS in 5m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5d7f4d31c6bc4d188b8af6c47bfa6f0c : SUCCESS in 5m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/78f9e384de9749289ae2e76b22b21b17 : SUCCESS in 4m 29s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d9cb44adeff444d3be6d7348a777273c : SUCCESS in 11m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/cfa185682112426397c6e86c54b320d2 : SUCCESS in 10m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/689bfbbd6488494b95194b39363f891d : SUCCESS in 10m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9296eb561dc84adabc8327137220c889 : SUCCESS in 6m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1547c8d5539e4b7eb53f5addeb961a5d : SUCCESS in 10m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2b95dcdc58234a8db76fa473108536da : SUCCESS in 9m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6de682a43f314facbbbfe99d4a25172e : SUCCESS in 11m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/46298c853119402a8141e3af9397675d : SUCCESS in 14m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fe15164158d842c18e4d6d02647fb88c : SUCCESS in 10m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/851ed976db69431fade7e4099cc044fa : SUCCESS in 9m 57s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7fbf3ff67e02413488d3f2c666af7b90 : SUCCESS in 27m 44s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/62ae535e2c17484ea68e858db2a40435 : SUCCESS in 13m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/78f1882db1ff4766a4f5137ae996994b : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c89e5e4e5b15488994770833c9d188a0 : SUCCESS in 26m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a8fee6d95dc41109ac423c7f6147ed2 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6c5f1adee3b24556bbfcc23eb8b69d0c : SUCCESS in 8m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/be2c46891ed448c8970f2b9e9e74bb1a : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52e38e6597194f8290c92106111cb334 : SUCCESS in 46m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/81c7410715c4488981fa408389d67802 : SUCCESS in 1h 18m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6aec412d86169d1a3c55aac63ebec4703475a0d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:14:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9a62bd7b962bbf2069283067e016f52a8c5c4d15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:15:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21c2cd8497004d01a1268251a87488b8 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5930f2fbd44667943b9ea20fcaf995f25dde4954","revisions":{"90b07f71202de058a6cb596fa97199e3afbf244f":{"kind":"REWORK","_number":1,"created":"2021-06-08 10:17:56.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/70/795270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/795270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/795270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/795270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/795270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/795270/1"}}},"commit":{"parents":[{"commit":"748e981ff163732e486e7f0d589e4907fdd52454","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/748e981ff163732e486e7f0d589e4907fdd52454"}]}],"author":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-08 10:04:56.000000000","tz":120},"committer":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-08 10:17:53.000000000","tz":120},"subject":"Add ruamel.yaml to global-requirements.txt","message":"Add ruamel.yaml to global-requirements.txt\n\nThis Python library is now a requirement for building\ndocumentation (ansible-autodoc) for the following project:\n- openstack/tripleo-validations\n- openstack/validations-common\n\nruamel.yaml is a YAML 1.2 loader/dumper package for Python.\n\nThe library is actively maintained, with almost a weekly release. It is\ncompatible with Python 2 and 3 and licensed under the MIT.\n\nhttps://pypi.org/project/ruamel.yaml/\n\nSigned-off-by: Gael Chamoulaud (Strider) \u003cgchamoul@redhat.com\u003e\nChange-Id: I208b00629493953a502f8b95756820ae0d9a9cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90b07f71202de058a6cb596fa97199e3afbf244f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90b07f71202de058a6cb596fa97199e3afbf244f"}]},"branch":"refs/heads/master"},"5930f2fbd44667943b9ea20fcaf995f25dde4954":{"kind":"REWORK","_number":2,"created":"2021-06-08 17:38:47.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/70/795270/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/70/795270/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/795270/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/795270/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/70/795270/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/70/795270/2"}}},"commit":{"parents":[{"commit":"748e981ff163732e486e7f0d589e4907fdd52454","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/748e981ff163732e486e7f0d589e4907fdd52454"}]}],"author":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-08 10:04:56.000000000","tz":120},"committer":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-08 17:38:19.000000000","tz":120},"subject":"Add ruamel.yaml to global-requirements.txt","message":"Add ruamel.yaml to global-requirements.txt\n\nThis Python library is now a requirement for building\ndocumentation (ansible-autodoc) for the following project:\n- openstack/tripleo-validations\n- openstack/validations-common\n\nruamel.yaml is a YAML 1.2 loader/dumper package for Python.\n\nThe library is actively maintained, with almost a weekly release. It is\ncompatible with Python 2 and 3 and licensed under the MIT.\n\nhttps://pypi.org/project/ruamel.yaml/\n\nSigned-off-by: Gael Chamoulaud (Strider) \u003cgchamoul@redhat.com\u003e\nChange-Id: I208b00629493953a502f8b95756820ae0d9a9cb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5930f2fbd44667943b9ea20fcaf995f25dde4954"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5930f2fbd44667943b9ea20fcaf995f25dde4954"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
